)]}'
{"id":"openstack%2Frequirements~725824","triplet_id":"openstack%2Frequirements~master~Ie802f865ceacbecd5887817fd0f25137d28f5350","project":"openstack/requirements","branch":"master","topic":"influxdb","hashtags":[],"change_id":"Ie802f865ceacbecd5887817fd0f25137d28f5350","subject":"Block influxdb\u003d\u003d5.3.0","status":"MERGED","created":"2020-05-06 10:35:01.000000000","updated":"2020-05-07 11:59:52.000000000","submitted":"2020-05-07 11:58:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725824-1588852680613-452ece20","meta_rev_id":"0dba072045f74f33433b83d4b0ba43c6ad14c863","_number":725824,"virtual_id_number":725824,"owner":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-07 11:58:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"date":"2020-05-07 08:48:19.000000000","_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-05-06 15:48:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-05-06 16:02:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-06 14:53:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":1,"date":"2020-05-06 11:10:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-05-06 16:02:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"value":0,"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 11:10:19.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2020-05-06 14:53:06.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2020-05-06 15:48:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-06 16:02:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-07 11:58:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f7d6e50dad3814cb5a694617f23479d9ca9da8d","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-05-06 10:35:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f598da97e334f370aa97c1e183069a7162e7f79","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-05-06 11:10:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0ccfdecb2074f4df9d7388dfb3c8e751324a10e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 12:08:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d2410d4a0df46e0b6e2fa4ae237fe46 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab58eab0eb0b4a0694b7559caa078d7e : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09e3688f5e9a400588f7a378b4a6c9bf : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bf49d764fae47c0b639d3ee3c082778 : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ea43386304d48048d62cbec954f9300 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44bfc1c37e70400d8c8099290dabfa99 : SUCCESS in 4m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0ae351b2e33f471c969b513e52d298ac : SUCCESS in 8m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b8ace08549bf49339e5d4de3fb3208b6 : SUCCESS in 9m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aea3eca88d734ae18396c85ff6c92f01 : SUCCESS in 5m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b34dc9cb3aec4784bea068a3df3ada8c : SUCCESS in 11m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bec7d6463c4c466f8c7e33731f0ccd3e : SUCCESS in 6m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e127c254fff4460ab9da5aaa963a153a : SUCCESS in 13m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f5c00851451447d9b01c24c0ae1bc091 : SUCCESS in 27m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/42b508a8c54c435fb917f2b25d365d0c : SUCCESS in 11m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3fadb84b505148b3a22b03d2e7018668 : SUCCESS in 13m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a60c6bbf44e14cccbeaafaec7e0768bd : SUCCESS in 15m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/573f8fa83fe5408bb803258de4222630 : SUCCESS in 6m 35s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/190234b8b16b415b9ae9c193d24dccff : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/454a6b96dd764938b75d5e0ed9993427 : SUCCESS in 7m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a16512fce6ef4043973e1ce51d20c811 : SUCCESS in 10m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/152b3aa0ce6040ac8bff9f5ccf8b9390 : SUCCESS in 21m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/109a389395374b98af0eb8c7afd69f3d : SUCCESS in 59m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8da1cfcc5494a03904ade5ef744088b : SUCCESS in 1h 07m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ae5afe3dffe4b218b1250057c992062 : SUCCESS in 1h 28m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/513ebf03fb2d493789d3f3f4b383f708 : SUCCESS in 1h 03m 43s","accounts_in_message":[],"_revision_number":1},{"id":"41ebe6cf8aedfbc750783867ca2fb37908e82ee3","author":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"date":"2020-05-06 14:53:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"95cdb3868b30dc2419d3b06f0a09048fcd037937","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-06 15:48:49.000000000","message":"Patch Set 1: Code-Review+2\n\nnot quite ack\u0027d upstream, but enough other people report it...","accounts_in_message":[],"_revision_number":1},{"id":"0c3426c4c407edc2a4f826ed30ae2342c40e0180","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-06 16:02:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c1cdae18dc073827eb0db604d6fa5875298684f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 16:02:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fbdd0cb27ecbc3753619c3cda79cdf5f920b267d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 19:37:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/399173b438b74efeb47025d91a663a27 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d8b2a1241af491b8fe613e41ad6d104 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29ba1e4a95bc4589b1c74d900ea847ce : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c20760168f5b4d61ab0a88dbf11e0f11 : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/25996093c84e44f1af505ac865f5ea45 : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f17ab0b987374ea38e2f5be0b9856ee5 : SUCCESS in 4m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ae9b41c278ba4342bac9bd416d2c3788 : SUCCESS in 10m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c8f505a22c844cfcae46489f338c4221 : SUCCESS in 10m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/50dba943d7b945d3931b7fe58badf73c : SUCCESS in 5m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/327f97d629714a02bf3e81eb61eaeaa7 : SUCCESS in 12m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b175f715119c4954b3b18ec2a04b98e8 : SUCCESS in 9m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a7e29823097a41f7b4ffc966fdade332 : SUCCESS in 14m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9cb89bc97c944b40af5265354fb4d9d8 : SUCCESS in 26m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f9fe8c31b5c3425296f565fc62a3359e : POST_FAILURE in 18m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e99ef8a794fb4819930259f4ab0c657e : SUCCESS in 16m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a26056bc6a1748399507ca40963a7391 : SUCCESS in 19m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ff58c0e1a84543e487e8ebc5d52f7fdc : SUCCESS in 5m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1872d89dbfe0419793868697ee554b11 : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8ced90adc75c4d698052f99deec39be6 : SUCCESS in 6m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/090be3b03e604069be29eeaae18ba7bf : SUCCESS in 10m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/75b397ebbf6f42fdb1e24227d348d791 : SUCCESS in 16m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b367fe32a954598bc286fa30b91c905 : SUCCESS in 55m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/346a49590c3744c089fec11a78f5660a : SUCCESS in 1h 11m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f8bdbf58f904fc79e868a30ba3c1c6d : SUCCESS in 1h 23m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d35189148e22493fba0b6d20f9e540b9 : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9e37e1c46a7c55f6822b60850435b230e43a18fe","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-06 19:38:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d37c5d5d22cbaac8d2d94abe73347d3fa877661c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 21:13:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb1d982e109749aa8d65684ff2aca480 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/753b1d726fd14f6297774f9c82e0c37a : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34a6edb07e424160860b9ba987cc1d8f : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcdda0ffeec544d389f099d6cb187114 : SUCCESS in 8m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/22e4ab3aac1f4d2f8ea2f919dd839dc4 : SUCCESS in 5m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0c180171f85452eb4c0ac65be3ff123 : SUCCESS in 5m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bab10934a28b43a583f7a0ed3d9ec9f4 : SUCCESS in 9m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b4586bf0181043129309ce2349616812 : SUCCESS in 11m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f1077c85033b4bfaa6fc3c9bc1e9e316 : SUCCESS in 8m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4554e72f70c244a0907ba97c6349c5d2 : SUCCESS in 15m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/65f25b92faf34af09556c20b51f2b08d : SUCCESS in 13m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3096da051754466185bf6c1ad0df82e2 : SUCCESS in 21m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/16f13b1e6fb840559fedcbcb77e23174 : SUCCESS in 31m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ebbe764b595e46e78ae19cbfbf7429ab : SUCCESS in 13m 55s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/40c2d6d2cc5b48c59ce93c1ca0f45818 : POST_FAILURE in 21m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e376b32ade241d1b36edcc3dd73fa4b : SUCCESS in 18m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7ce76f56baa9464fb09c5a8aad12ed69 : SUCCESS in 10m 15s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b3fba9486bb34b2786d626fc40418dc7 : SUCCESS in 8m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/03abc2d1491d4ef5bcc70e69f505a74c : SUCCESS in 11m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c5422a2d2978476695c509535d238903 : SUCCESS in 15m 52s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ec249fa9035049f5b8bf96466ac988db : SUCCESS in 12m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b36eff18a12455ab0e932d5f14ba24c : SUCCESS in 53m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/062c1f2b437a4a2f8c52bbfad1dff4a2 : SUCCESS in 1h 15m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83780b14550c4588be02ba0928784cd6 : SUCCESS in 1h 25m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/14f0ebb2786f498ba6e6851720a31342 : SUCCESS in 1h 03m 45s","accounts_in_message":[],"_revision_number":1},{"id":"218b37ba8d39e7ba076499a01b35ad4c05bc3ecb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-07 01:59:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"09bc6145b95d98cb39cac0c9c97dac0ebc48a1e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 03:23:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1c09b16a41a45fdb335c719aeef3e80 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0953af83f86146f98bf5798ef49fb53b : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa74bff5e78748fdb05efd76549db30e : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/799a06e0771b45ea96a9dd6593ca31d7 : SUCCESS in 6m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a6e2467a4444b708363cd490c1f76d4 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/97fbd523f8f54766b0588e7fd707e86e : SUCCESS in 5m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d92005ba833a49b0be69ac9fddf0ddc7 : SUCCESS in 8m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5e4a623dd9004e798c12b44ae0dd6d3d : SUCCESS in 14m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/538a772f53ec42959a950690924749f9 : SUCCESS in 4m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/07a8e50c3a574bbfa04c44891eafd8e4 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/92d942975bdd4b9d92dc36b6fbdf3fbe : SUCCESS in 6m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/46b0a54c45184bec99214740d6e3d040 : SUCCESS in 40m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8e537a0faf534df1a8907263159f1f5e : SUCCESS in 26m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e3354c6f44da4b0a9e032303adc1d983 : SUCCESS in 22m 03s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b4220547ea084cbbb9af6264d4eb9d69 : SUCCESS in 12m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7009bcbb00b246feafc1f04693b0bc33 : SUCCESS in 21m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f054a4f363c4411aba6bbb890b1892ee : SUCCESS in 4m 38s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5665bc208fc04bd5bbf3df1f843b5a8d : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c9922d81adf4792b459e2157ee4fffc : SUCCESS in 6m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b66cffee2944460ebf60f435c57e7f2c : SUCCESS in 10m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2ff1bb07aeb345f1918d11d226ad3a98 : SUCCESS in 18m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c66d74bccdb4e06b3639b05ec4bbeac : SUCCESS in 51m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cc6340ea5764457482e588c542b4c838 : SUCCESS in 1h 01m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9417d5ab0c1742cfa70054d3304e1695 : SUCCESS in 1h 19m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/e47125aeb5af4a6187a83c243cffcfb1 : SUCCESS in 1h 10m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5438b470e1f5629e4fc08f75510a5409ce325cb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 03:23:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64451a20a4e8ca23ee4517a267fe79a0e4f24248","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 04:49:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11ecf0978f414b33ba748ed1b4afb830 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad572291951f413db3de6f8d6280673f : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b6c9c882ad44de5a2f925bca508ea06 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a1fb349aa064f5d8a978f6d46a80a78 : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a306b3f5fe564f4b86c601a82a68f951 : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3949f8492a64352bdd116c772f70b6d : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9bfa5670897548a2a37c81535753efaa : SUCCESS in 8m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2f2965b1e90d48139959defe11e32a4f : SUCCESS in 8m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f84e1674d9fe43ef83cb7fd09ce85b51 : SUCCESS in 5m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/091d5fd5a1b34fcfa7c7b5af9593bb22 : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0ce11341aeb54cd7833bc50107b6a93e : SUCCESS in 7m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b80d860abab04c79b14c7c4736fd6907 : SUCCESS in 18m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1ababac5c98041b6a7e19b89fb576a0c : SUCCESS in 23m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ce8863f83f634598b5c92fc73cb67467 : SUCCESS in 14m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ef8c4021ef454414ba3913187ae785b8 : POST_FAILURE in 13m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/68b7e9e105c54e579431b9de6ae6f707 : SUCCESS in 15m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d4ff0529120c4a80989fed54147fddcc : SUCCESS in 5m 30s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/fa0096a11ae44110a29d457434ce5d72 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ade5969f1bbb479198bd4cbf5d223fa2 : SUCCESS in 6m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a39842981a594531ae81572a4f0f002c : SUCCESS in 11m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/75757c1eff054153980e28308d490ebf : SUCCESS in 10m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87cc9bdd83bc430db6b3adc37a0269a3 : SUCCESS in 50m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2219c912eb9e417b86c63a8c00a8387a : SUCCESS in 1h 04m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/548cb5f5575a4bee9825e92c9001ac82 : SUCCESS in 1h 16m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/5491436a7c5749f89dba83c8d1c8c9c1 : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a28ed51010b93908387aef45175ed530f19985a2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-07 05:01:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de2e246e113ac6f1e966c87497f2f1da4bc77c64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 07:01:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68a69efbda804afa8cb0811a4c258ee0 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b72d3091fd441f49bac668f13228ebf : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e2882bebad94a0b99f6de71027d3fd8 : SUCCESS in 8m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2058fb5d7c8a43b99697cf140773c803 : SUCCESS in 8m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c7e9160774f4ac7bba97467747abe7e : SUCCESS in 7m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0a4ffc8a948b429f96f40e70424f434f : SUCCESS in 3m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ce6398e8ce9c48f993b5d6d6cd3daeed : SUCCESS in 14m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e23ef561be5f46548404b43d79cdb4a3 : SUCCESS in 8m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0bfaee2ade4c40ce978fbf4de2e5f5f6 : SUCCESS in 4m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/36cf94492d7848a99f5fca3542ce6779 : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/47d01ed2f9594036a6c5398225705069 : SUCCESS in 6m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cac01f9d73804fdbadcc96cfd4c460c9 : SUCCESS in 38m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9ac5e616e08545529ec1a071c3a7fb0e : SUCCESS in 24m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/26c126db638341b79234d47c28ee8290 : SUCCESS in 22m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/18ef3a9545f04a5db749f563f85d881a : SUCCESS in 19m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1caab7c351d94a2ba60e1d0b89359736 : SUCCESS in 22m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/208681fa630e456ca08f5b45c5bfec94 : SUCCESS in 5m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/c8a316f3677547239f88b33b5b2c0fe8 : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c914ecf2c31d4e80a83ef1f82869ccb7 : SUCCESS in 8m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0ae81e4a36174ff1a4f4f5b93d26cfc2 : SUCCESS in 19m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d7647febc6de4eccb3e0a201be565f33 : SUCCESS in 19m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e271c8c817184748bd2721548399cef6 : SUCCESS in 44m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/76e2230f8c244e039e68746e379852bd : SUCCESS in 1h 17m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5891910934646f0981c6a95289d8c17 : SUCCESS in 1h 58m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/dd01488a5d404d17b6c2391a32ccf15c : SUCCESS in 1h 08m 09s","accounts_in_message":[],"_revision_number":1},{"id":"04ea5d3423774eacaf4cf3ec66f0401f5b8002b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 07:01:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a4180855ae34c34f086aa139553d93fdc55fa6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 08:36:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/569ef77587ef4a37ba2ce0f2881e831b : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/176355dfa4df4f78bd137f37310aee63 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd315a11af3c4891ac42bd07af70dd33 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6d27aa15e18406cb71482effb7a47c4 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/05cc81075e5546008041a2f05c8d4735 : SUCCESS in 6m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/62a97f88153048069615ceedaa4427ec : SUCCESS in 4m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f1bcbe84026f42bc9f728fbe1d0684a6 : SUCCESS in 10m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0786e7a1ca0a4d5b863a98e740fe9c47 : SUCCESS in 11m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ea5645e1be1b48ec963d8dc57864830e : SUCCESS in 6m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/55bb2c6a97e8410da870712af0990fe1 : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1ef8666c005f485ca779db7ad434ae81 : SUCCESS in 9m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7e92674c25f44dc9af86d5813a34f115 : SUCCESS in 25m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0bdd4c62826b44948359bfbe90ee166d : SUCCESS in 25m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/28057b9134454995978433262ec5429d : SUCCESS in 15m 08s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/bdf484034acd4dc681e366901c2e94b6 : SUCCESS in 17m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/714c6be1d06d4385b3777992c50e68d5 : FAILURE in 21m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/de8072e302c448a39d00e890ce3228dd : SUCCESS in 5m 34s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/183239ff239847c5bfb8da7b2726c566 : SUCCESS in 9m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80aee48c1b32411bbcf2e60b64798dbe : SUCCESS in 10m 52s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cc44ee71b1e3435a9f7f73b640787f49 : SUCCESS in 23m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5d85e3209efa4bdd951380d94bddc7ec : SUCCESS in 23m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf5ad5567e95472d8295b4b738ebc7b6 : SUCCESS in 59m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/65d4deb190034285bd55da5f4da8072d : SUCCESS in 1h 07m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32a943043cb844c0b5ade3346a3c3bd8 : SUCCESS in 1h 28m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/51d10300c18f4718b72f2c32cebdb599 : SUCCESS in 1h 08m 39s","accounts_in_message":[],"_revision_number":1},{"id":"1f6c1e5f616d66dd25d8058cccf326aefd38403e","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-05-07 08:48:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"86e92d5eac6c6b91e80e66589fa347908dadf88f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 10:18:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f5b7110c1c4f3f8ac70d91288d28da : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce7d2a8a0dd7427f93fad4ebbde8bd62 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/583b73334d0b46a19f261daea2dab05f : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51c0001c356641cb9008791bb62e2b41 : SUCCESS in 7m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bf8f06f15de04c0fbda5ac2c8105e2bf : SUCCESS in 6m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e7929b975b7e489388e8c75f04fd34a0 : SUCCESS in 7m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7350b254cc4a47e8a5b56390ae5f38bf : SUCCESS in 13m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/afb56cbe28514d9292d593ffda726ff5 : SUCCESS in 12m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b4e280d61ac34653b5a15b545faad5e3 : SUCCESS in 8m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/61f68361344542fba4a980ec94266ef6 : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/17a8acd7cca843589bffef0156bdef4b : SUCCESS in 12m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c01dab5f7cff4390a17199fde22df6cd : SUCCESS in 25m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b1ce7a20a2554153b91e363acac271b6 : SUCCESS in 31m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1d202551a02a4f5b87b75f4dca11160c : SUCCESS in 19m 15s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/7c73226d8dbc4420936091d6eb91c99c : SUCCESS in 21m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e2ae86a0560e42a1b2f8813df71b4775 : SUCCESS in 25m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3e9c3f2f96424ee39e45c1ad4e519f77 : SUCCESS in 6m 51s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1e18cf19ec2b47daaddf83e5b50e0e2b : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6ae911455623444da6de078cdbded7d3 : SUCCESS in 11m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1f30d861136649b5af87cd9d3133a808 : SUCCESS in 10m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0333daf1e2f14b519a839f6abf2722d4 : SUCCESS in 21m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7aa9f03ab4114d4a9bacae3b2fabe6f1 : SUCCESS in 1h 02m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/475ddc2579614d9089b7b9e869f88da5 : SUCCESS in 1h 19m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/549a2460fdba4563a53af835bb42bdaa : SUCCESS in 1h 28m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/37de384902784e54add0c35af9417fe9 : SUCCESS in 1h 03m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d021d0600d7379a0b9a5d63b54dd39cbda4e22f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 10:18:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7bebaff8661d5c8ab7607762973cca4670bb4268","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 11:58:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3224eee5d3544f1896fa069e638b47d1 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ae4646ffd5945e9a52a0a6c1cf1b83c : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3eece4bdb12468cbcc195c4a6f17394 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91018277ec1c4e84a7883583b4adfdb7 : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da25784d437c434c8d4f02d47d45996b : SUCCESS in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eaac845e18ba4b30988e113342804ffc : SUCCESS in 4m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/015f43c2be1d485ab429c0d7e9262878 : SUCCESS in 8m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a5514d7a7a254a07b2e394ef3536d0b8 : SUCCESS in 10m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e77b3ce0e4a440c8be5a541b1fc6b5db : SUCCESS in 6m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/55d92aa78f5e4f368cc9b52d84783438 : SUCCESS in 10m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/76d8d29f3b474e88ba76d8dd4638536b : SUCCESS in 9m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/db1a07aa2c0f48deb54c1fad09392670 : SUCCESS in 13m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5f6a37d57ea944c1bb405397a2b26129 : SUCCESS in 31m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c9eea7caf3c94206a4311af5f8acb42b : SUCCESS in 13m 45s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/69533ae9c561406bb7e749991d71c4a8 : SUCCESS in 14m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78fa1969717648a2b32060bf9a1f04d1 : SUCCESS in 18m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d59ab1bd72fc4a5289c80d44aa46fcfa : SUCCESS in 5m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b0a78158f47240c99695ed6dd8fb8607 : SUCCESS in 6m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d1a13ed912a84050bd75f1390fdc0764 : SUCCESS in 7m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5cfb4a61c2bd4a98b495b7d56a64d9fa : SUCCESS in 18m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/90759a78c02648a9a4fd05ce3ed03721 : SUCCESS in 11m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c40849f944a3445ba5771d8b32b3b95e : SUCCESS in 57m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e73faf24bbb496a8a8886b50d4c3d5d : SUCCESS in 1h 22m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0ab3cb3be9446dcb84e35c5aa947e25 : SUCCESS in 1h 32m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/dfae46ec212047559e9802c3fbe42996 : SUCCESS in 1h 07m 53s","accounts_in_message":[],"_revision_number":1},{"id":"bc759b055bc069833f034080a6954c5361a070cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 11:58:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2c91d9176d81d9a048d0c61e4d8897fb5891811f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 11:59:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d6bdc9442d4d789da48150a7c1d1a8 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"981369e9b729e0143d74c0c9a3b2ec726c582aff","revisions":{"981369e9b729e0143d74c0c9a3b2ec726c582aff":{"kind":"REWORK","_number":1,"created":"2020-05-06 10:35:01.000000000","uploader":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"ref":"refs/changes/24/725824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/725824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/725824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/725824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/725824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/725824/1"}}},"commit":{"parents":[{"commit":"a7899d17f3e61a1b4903f957ad5c826552d9f1a6","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7899d17f3e61a1b4903f957ad5c826552d9f1a6"}]}],"author":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-05-06 10:26:37.000000000","tz":120},"committer":{"name":"Martin Chacon Piza","email":"MartinDavid.ChaconPiza1@est.fujitsu.com","date":"2020-05-06 10:34:07.000000000","tz":120},"subject":"Block influxdb\u003d\u003d5.3.0","message":"Block influxdb\u003d\u003d5.3.0\n\ninfluxdb release 5.3.0 includes an open bug that breaks many of the\ncalls to influxdb. Like for example \u0027SHOW DIAGNOSTICS\u0027 [1]\n\n[1] https://github.com/influxdata/influxdb-python/issues/820\n\nChange-Id: Ie802f865ceacbecd5887817fd0f25137d28f5350\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/981369e9b729e0143d74c0c9a3b2ec726c582aff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/981369e9b729e0143d74c0c9a3b2ec726c582aff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
