)]}'
{"id":"openstack%2Frequirements~726072","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Ib8fe526226b93eecbf2dcddc0c2727ad08c07b3e","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Ib8fe526226b93eecbf2dcddc0c2727ad08c07b3e","subject":"update constraint for paunch to new release 5.3.2","status":"MERGED","created":"2020-05-07 09:39:58.000000000","updated":"2020-05-08 04:27:18.000000000","submitted":"2020-05-08 04:27:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726072-1588912038588-155cca4d","meta_rev_id":"ad65d6708a7f916e12c0ed9ff8d7c8fb5c21dc9f","_number":726072,"virtual_id_number":726072,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-08 04:27:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-07 12:55:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-07 17:45:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-07 17:45:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-07 12:55:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-07 17:45:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-08 04:27:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0eb7ffa979faf53656e88c75146530da738ea0b2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-07 09:39:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10e98784b8cf26c4e8b66733774beeedfae99e31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 11:20:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/aefd50bb982941feb1ebfd4ea2bf9783 : FAILURE in 4m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bea7cf3fa65147d799f9635ce5f9331e : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c556c8881f24706a8c6c84706131b20 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d51ac9e2c14140659b2991ff6857a228 : SUCCESS in 5m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f35482b6bf843bd9678ffc96688ad57 : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76ce4a0633e246c2a16b3ee117c8cf79 : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b759b242428e466b991161c66791c5c6 : SUCCESS in 4m 34s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/d003803664414980ba2fa176d4596149 : SUCCESS in 8m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/55f00065930d43ffb602edcd7a308986 : SUCCESS in 9m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/85b33874d5a24a0f8bb72d580d520199 : SUCCESS in 9m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9c8229233df443248e6a859ba0e343d1 : SUCCESS in 10m 25s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/47b4d6e5a79c40ddb2e3c6ff9299dfdd : SUCCESS in 5m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/50bfb9dee6c249e4bd9192885006f8a2 : SUCCESS in 5m 56s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5557ff0dbaf642458b94b06a4fc34a7b : SUCCESS in 11m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e21a79c38766445eaccb8e23c36d56a9 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86bcb2e486904d7293cf64b5ae971dd0 : SUCCESS in 9m 29s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/90a74b7f66d34290900ff309f2b4e108 : SUCCESS in 16m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2eee6cfabc2744e8834be15ca06612a6 : SUCCESS in 13m 50s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b0b847c4c0194ac89f0a02eb98be3f77 : SUCCESS in 21m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cba42ebfa0464585a40a7d2b0ee409ea : SUCCESS in 23m 59s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/48112680c22646308be1490b9ccc6d0a : SUCCESS in 16m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b607db529b92436a8cdd43892468b8f3 : SUCCESS in 12m 00s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5c84811457514350a06a53ac88829c03 : SUCCESS in 18m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e4b9846446a42eb8afd2736c97e38d8 : SUCCESS in 26m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/65b06866aa784a0c99bb3f1b132a2463 : SUCCESS in 6m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0e743e27529148e6b7a8b8727e8be3a0 : SUCCESS in 10m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a1b624b13a484e49a8298027047340bd : SUCCESS in 10m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bef99f849ad14c079051a09471e92929 : SUCCESS in 1h 21m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/397b42bcf0344319b56debcc4a3f2a2d : SUCCESS in 1h 35m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4aa6d6a3596b4198ac479fe6faffa003 : SUCCESS in 1h 08m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2d2f071acc66143d1b10b5f936de973eac5a8e33","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-07 12:55:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b1c3bea8413336b3f33fe32fe28434c54cc67cf0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-07 17:45:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e5ecbf4e6ebbf24169054986774e85817db42974","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 17:46:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6bb783d8d5551a83816f02b2de53501abac83582","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 22:58:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7094292e2a8d45fb990d69e87555e440 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/25cde5cf3e9944e6a56fef3b6c7f1153 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00dd3092fde7455f984c6e36fd3bc48f : SUCCESS in 6m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c66be17f24314615b864152e47424a11 : SUCCESS in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/215ad8fadf6c49049b5e8c18b17d103f : SUCCESS in 5m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/462f9e88009a41ae9e5435280bdde5d7 : SUCCESS in 6m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/46b702db96c84940b07b0bb866157678 : SUCCESS in 12m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/def0256af3a146b5818ec9a1d018307e : SUCCESS in 9m 25s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/15cf9b83d77c465eae3a221ca5a23870 : SUCCESS in 11m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a231d1fbda0b4e49a667440c5c4a9cff : SUCCESS in 9m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3aefb29db82541b28a325250d97ff355 : SUCCESS in 6m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f84fb557aa5c49d1974e927298b411ae : SUCCESS in 6m 24s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b86889487dc540c1be428fd5bc3bf0a2 : SUCCESS in 12m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8ab118da16474cc9b97c41bce6432543 : SUCCESS in 8m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/321223e7b6eb4e5e98c3b8dcb91b516a : SUCCESS in 7m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/39f6e5fe7c8a4e4e980f8c5f17e6d4a4 : SUCCESS in 27m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6cd8e5ef44e648cdb085f187e3838c36 : SUCCESS in 16m 22s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8be94dc85213421f8f244d73d09fd7e2 : SUCCESS in 22m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0a41cb31aa3a47ad86c9b77144180cff : SUCCESS in 20m 41s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/10d90c98aa5342de9b042001a9060cc7 : SUCCESS in 13m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a36b17f5f5c1487abcfe01a8948fc6f7 : SUCCESS in 13m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/ffd1e05177584b58a3d6e53e155285d2 : SUCCESS in 15m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2676f986bfd140329ab1b77f077f1a30 : SUCCESS in 18m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/99d0ba4d815c493a9d73ee80b550cb61 : SUCCESS in 7m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fbf25a3d2cfe44ec8358507cb2d27851 : SUCCESS in 12m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4f6d4b271eec45b4885e03e6703f9d31 : SUCCESS in 14m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d283455560424aea8d3d074b5d225067 : SUCCESS in 1h 05m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6789dd0324374c7daa3eef9dda2b4040 : FAILURE in 1h 16m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7c370c8e73ac47b49293c2922e290265 : SUCCESS in 51m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d91d52e25e13f09967467075a11351234c2c7abd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-07 23:10:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"62112eabe0571b61891c5a3f364ec7e6d2323237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 00:44:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/c392e5a66d844c3e8ac13223ebaaf397 : FAILURE in 3m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/232b3c02c63d44c781e17c9663885588 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d54d3f98f8e4018a80bf3ecec399766 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c67a4995d8b74c9f887f53feaa6f8784 : SUCCESS in 6m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f8225c056a744e5b33e0bf60d025f87 : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/662bc0edaeb6455b8e2e772f3a1bf93d : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e86be7ce67b4495b0ff0c4ce4d45a9b : SUCCESS in 4m 27s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7541feb6a49e4dc9b3f19dc49059039f : SUCCESS in 11m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7ae3277bd2784a1b8e97eeaec02a7a23 : SUCCESS in 9m 09s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a3734879306a4a95845261baf90e9c15 : SUCCESS in 13m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a5d503105ade4c90aca1151944576635 : SUCCESS in 10m 14s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cb8b5a32f31d4f2faddd1af14c7fdca4 : SUCCESS in 7m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4eb9d16eedb1445ca79020230a0f37dc : SUCCESS in 5m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/eda4281571474587a397371619624ab8 : SUCCESS in 10m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/18f68fa29a3f48dab2dd60710eafd5a4 : SUCCESS in 9m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df2e698597d64f82b466f1810a7d38ca : SUCCESS in 9m 34s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2689d6c5c64b42f0a8ebab72975b524e : SUCCESS in 26m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/769275d1f0db42b6b2f0f6584b4c8961 : SUCCESS in 41m 01s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f670034fc12e4f2a8db532f760015ac3 : SUCCESS in 25m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/367b9ca6db954c7c8240de7071f8feda : SUCCESS in 25m 54s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/924006571c32429f858ed013bce17838 : SUCCESS in 27m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e96edfe554f446a4bec10187921f35ba : SUCCESS in 26m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/59fd936e64294bb3be0d77c954f40686 : SUCCESS in 20m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/25a59d8847f747cbac68142dfa844c8e : SUCCESS in 22m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c3946605d20e4d71b7e18140f13ad4bb : SUCCESS in 6m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/774a055ef31c4577bd978f6e766cab51 : SUCCESS in 12m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9f69491a24b04137a58cfbe7dd190685 : SUCCESS in 10m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e182cdb321c6418b881520b3edcafe11 : SUCCESS in 1h 18m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/65b42d8e598548fc85a0d0fb51bd8611 : SUCCESS in 1h 33m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6ad9bfb6cbb0460e869e0555b43d4700 : SUCCESS in 58m 45s","accounts_in_message":[],"_revision_number":1},{"id":"9717fd590cdfe32e348fcda00974f51f6cbff325","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 00:45:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bcd085a183af20e62cd4d428abc583a21f73373c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 04:27:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e15246c545154837aefe3d84285200f2 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad57b5b1df8d45e89a95933cdb745d94 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7d2d606c864405aa06743fe4f9dcb7b : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5160cedbc7ab4e4f8a3aa9863b05a250 : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/143c7cabc7af4a5dae0bdc9ad4a02735 : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/03c4c068647b4a15bfaabbbd37fdbd60 : SUCCESS in 4m 16s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/28e74e7422e74cc9908b8d53a1e398d5 : SUCCESS in 7m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/defe2b0637a14fa0af0501b309ed2a49 : SUCCESS in 9m 29s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e3a228f79b7947caa101d3740601a66a : SUCCESS in 10m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ee20965be85e48579e9eb0230b8e6b48 : SUCCESS in 8m 40s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d2ab7c571492490385dc438ee949e895 : SUCCESS in 5m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7e2fd7a6dae74a42bba820990393bf6a : SUCCESS in 5m 48s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/85c77af033c3402397bbdb04bc54a2e5 : SUCCESS in 10m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1c2af416be1642d59b0d20985fdbaf4e : SUCCESS in 8m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/14e078b1614a4f34a972b63e83120cdd : SUCCESS in 8m 32s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/6b806e244d934dbaa54ec63fc1ee055b : SUCCESS in 13m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c3aaececce154e28915ea03496e43260 : SUCCESS in 12m 32s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/032263ee37664795b133a744e9b1b330 : SUCCESS in 25m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f63c1a339aba4f728ccd67356cf50e4f : SUCCESS in 24m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/92f352e97a634159a6cbd051153e0b4d : SUCCESS in 11m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7dbdfeb4bbda46de9d27337b43a57da9 : SUCCESS in 12m 39s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8f9cd24163734f0b9ae6276f7c434017 : SUCCESS in 14m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/10cc6501c9cf43169339c22d4a5fe575 : SUCCESS in 18m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f726d5894bb84823a23683e5275b54c8 : SUCCESS in 6m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/890939cf7cef45dc81c32f32caa87a03 : SUCCESS in 8m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/19c15096899e43a994903047081bb6f9 : SUCCESS in 11m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a357ac396eb84461976b1178bb0f38ce : SUCCESS in 1h 10m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e4dcff7f88bd4102837557c9c41aa756 : SUCCESS in 1h 52m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e0c1b91414a4486c8103a19829cb41e2 : SUCCESS in 54m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ffa83042fa06be7834c8f34e361c07a4ac6920d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 04:27:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef36a239ea199b9deb81ceb081068f980f23693b","revisions":{"ef36a239ea199b9deb81ceb081068f980f23693b":{"kind":"REWORK","_number":1,"created":"2020-05-07 09:39:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/726072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/726072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/726072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/726072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/726072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/726072/1"}}},"commit":{"parents":[{"commit":"5b7795f1c11c6294a75cfa0e53d511a2242d1163","subject":"update constraint for openstacksdk to new release 0.36.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b7795f1c11c6294a75cfa0e53d511a2242d1163"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-07 09:39:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-07 09:39:57.000000000","tz":0},"subject":"update constraint for paunch to new release 5.3.2","message":"update constraint for paunch to new release 5.3.2\n\nChange-Id: Ib8fe526226b93eecbf2dcddc0c2727ad08c07b3e\nmeta:version: 5.3.2\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Rabi Mishra \u003cramishra@redhat.com\u003e\nmeta:release:Commit: Rabi Mishra \u003cramishra@redhat.com\u003e\nmeta:release:Change-Id: I1ab7d717bf482828ccf997aa72da2b0d858b80a3\nmeta:release:Code-Review+1: Kevin Carter (cloudnull) \u003ckevin@cloudnull.com\u003e\nmeta:release:Code-Review+1: Luke Short \u003cekultails@gmail.com\u003e\nmeta:release:Code-Review+1: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef36a239ea199b9deb81ceb081068f980f23693b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef36a239ea199b9deb81ceb081068f980f23693b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
