)]}'
{"id":"openstack%2Frequirements~726320","triplet_id":"openstack%2Frequirements~stable%2Fstein~I0f4042e5bb17fde2c861a1e2c086d0d51d126588","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I0f4042e5bb17fde2c861a1e2c086d0d51d126588","subject":"update constraint for oslo.context to new release 2.22.2","status":"MERGED","created":"2020-05-08 09:45:23.000000000","updated":"2020-05-14 18:35:28.000000000","submitted":"2020-05-14 18:35:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726320-1589481328028-d6aca499","meta_rev_id":"d539e321efb730fc7cb311c7f14b393b652d1c0b","_number":726320,"virtual_id_number":726320,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-14 18:35:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-09 12:49:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-12 00:11:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-12 00:11:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-09 12:49:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-12 00:11:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-14 18:35:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac02aa57d01000c67f4e659f2c307eae62cabecc","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-08 09:45:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"919b59b88a58b7946cd18a029cb046e11f4bda1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 11:14:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/78aaf654a67641cfb7fd39e486dd5d20 : FAILURE in 5m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05c34aa265fc486d837b10ea58034948 : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/17df0c2177eb4edb9476f017b76fb8ab : SUCCESS in 6m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c54bc53a54404003920376faac29e10f : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a4f9abb268f49dab8abd75077892b56 : SUCCESS in 9m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8353ed990224588929fa8543f478d56 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a6df92bcb51046c19b65be771b25f1d2 : SUCCESS in 6m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3de8b834a7a14afe8483fc1b80a33a2c : SUCCESS in 9m 44s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/e8788c8bbda84f29bc16ae2d8e96981e : SUCCESS in 10m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7f99d774abf8414986c731b5e7cdabe8 : SUCCESS in 8m 54s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/682121b7c6294121ae8de7e103889028 : SUCCESS in 9m 40s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/a30405da4ee04159ba816babbac6f7e4 : SUCCESS in 10m 35s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ab1ca1d62ec44bc4ad06eb5bde12600a : SUCCESS in 6m 03s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/73abfb4b48db4401bf86cd69ebdbbee6 : SUCCESS in 6m 58s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ea24a3202b014f6e9651ac21144fc9be : SUCCESS in 12m 06s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/8ddf264d5e7b4db491aaa02265e78487 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/92a297cb792642c49a220edf55483755 : SUCCESS in 8m 52s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/87812a115693420eaad6a8469e85a3fd : SUCCESS in 19m 57s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/d18ae277bef643e9a1402e5f70d19494 : SUCCESS in 25m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/8b724710555c4e98a5d7ff76cdf03df1 : SUCCESS in 23m 20s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/324df1319c99450cacf8357e9c895ed9 : SUCCESS in 22m 16s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f34dfc59af0447e7b8d7839d673f403c : SUCCESS in 14m 23s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/eedefcfb7f6b4d67b5699e257567e3c5 : SUCCESS in 14m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/725c2c64be0944c084ca88dbc994953e : SUCCESS in 17m 29s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/2e5a3078d83845fba2403d024e4df40e : SUCCESS in 5m 37s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c1295d07afc746fc889b580916268d46 : SUCCESS in 10m 36s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/0111dcda55c44b56ae8ea81410853e44 : FAILURE in 4m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e29e1fc2d2f24b4c955603238be03302 : SUCCESS in 1h 23m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bb0feb7a721e47dfa17f8bf050e4d2d9 : SUCCESS in 1h 27m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b0e9fcb1e3564824a1344ea7f62d0e9a : SUCCESS in 51m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1517cdca6ada751fbb7bfe1729feb67f6af790bd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-09 12:49:17.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"15d64f16d8cd35d064bd7fd54ab55bbf4d355c8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 14:18:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/9d04b37f05aa4b349572d00a100aa3d0 : FAILURE in 3m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce729462fa594590982bd58f5c35ba24 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d8cb3056745e4138ad860d57e872ed96 : SUCCESS in 4m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a0c7ff69b4304057871a7f942b2aa083 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f0492058c18471788f10ba3cf9431df : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/713d50938a054be18029e4e2037c049a : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a5b7ad1695d1444bbda1aa4fd6ee41f7 : SUCCESS in 3m 16s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/665d6840296643eb99f4d341338e6dd5 : SUCCESS in 7m 47s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/e1facedc7f29426d93aacf76b49dc4ac : SUCCESS in 9m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f6ae241bf28f4a8ebf36df40ec78ab20 : SUCCESS in 8m 10s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/23c2acdbba4d4b26a6269477b99c5450 : SUCCESS in 9m 46s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/0554007ac84145aba1a64f530c0ad8f9 : SUCCESS in 11m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/9171e0734e5943e796347d4974dd0fd4 : SUCCESS in 4m 56s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/fe7c33a18c9f402ca38892a062335d99 : SUCCESS in 4m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/69581605a0f649cbb402b9d3ab9911f5 : SUCCESS in 11m 26s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/55f95fe07ac74fcd9d76cfd23061c00c : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ae04096edd94481cb9d9318119d78ca8 : SUCCESS in 7m 07s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4ddbc57593f14bdb9a2727faf6aff954 : SUCCESS in 21m 29s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/16f99b1b7e344028819ae9c9eb385487 : SUCCESS in 25m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f84eff6d57ed45d39fbc8a8191d9c65b : SUCCESS in 30m 05s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/5575dea30ba34598afd8e9d784563878 : SUCCESS in 32m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f2e4744a75774a2888acb98958519c0e : SUCCESS in 12m 12s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/d758145cfb8d413ea07b9e3a82978bf3 : SUCCESS in 18m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/51d4514b710a458dbc0af830a21cd485 : SUCCESS in 18m 56s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/0b2c726ec28543bb90406f2e275ab24e : SUCCESS in 4m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c20b324b060b423daf512b68836fc02d : SUCCESS in 8m 17s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/0c69f07b40de4908ad00a9266f15f7aa : SUCCESS in 7m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf940d3b2f0e4bd3b2da666335cd79f2 : SUCCESS in 1h 08m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/be700bb3eb2042988857c0daf483d442 : SUCCESS in 1h 24m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e048d3dab22d4788af42b3d0e1bec47e : SUCCESS in 53m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9e8482973bc9c0973a60d0acc926dded1d714954","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 00:11:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e5448ba4635a83420a5151ba565487594feb0162","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 00:11:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae2da62319d6a26c8c458771d14f96e94412e9a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 01:56:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ceb58bfd6d24779812d01be6e686297 : FAILURE in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f59474e0f9bc469ba0e9f3598276d592 : SUCCESS in 5m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5072503a25f541bc8960d8a1086e327f : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0abcdc97266c4e879f2402e035a4d9c2 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/573e1aa8b10f4ae2a5b8c10a6fb4bed3 : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/84917975532d4eb09e0439c2dd4358ad : SUCCESS in 6m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ab6393e5cbbe4aa384a6e8a4246da550 : SUCCESS in 9m 50s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/ccac8d3d9a3f4b9ba70dea112e210421 : SUCCESS in 9m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7187e5f665f744f0917e691408315864 : SUCCESS in 10m 44s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/62a0bb81f6fb4104959c9ea261ecd39d : SUCCESS in 12m 52s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1166ffd03d1f4ba0a20391105c108a0d : SUCCESS in 8m 05s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/c6a8556afdcc449b904699bf51004568 : SUCCESS in 5m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7ee4aeff1a9c4d42b1cabd4f111478ff : SUCCESS in 12m 38s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/6d73fb304c874dcf8e1d3e40bc840905 : SUCCESS in 11m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e2e89375c3844f299ec5b5f31cbe90ce : SUCCESS in 8m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/760e0bc0540c483c831db3a1aeaac5a9 : SUCCESS in 27m 01s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/b3d1945fc606428097c9afd5c6e6d81e : SUCCESS in 36m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3a345c8b5a8f44dc8f4829a1695ad975 : SUCCESS in 34m 01s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/fa850b9d69b2438090835e41f76f9fea : SUCCESS in 34m 12s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/b870ea02795845c2960d8224fc02c22c : SUCCESS in 15m 23s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/0870a94b17fc4498b638ad916469453c : SUCCESS in 13m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88513395442d47019e4bf266699c434f : SUCCESS in 26m 30s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/4b2879faf5d44aefb98f0b4b6ea1eb5d : SUCCESS in 7m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/43277659f7184565b3ad9a4dd8d189b2 : SUCCESS in 10m 00s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/794233c5a90c409e969d06583a30044e : SUCCESS in 9m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d6a2a5081e24f25adf71c945e5c0c3c : SUCCESS in 1h 06m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9db68c13818c4db3ad5f2714ecb83d0b : SUCCESS in 1h 38m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7d883cbe362744b48963b0b15b86965a : SUCCESS in 48m 33s\n\nWarning:\n  Comments left for invalid file openstack_requirements/tests/test_update.py\n  Comments left for invalid file openstack_requirements/cmds/generate.py","accounts_in_message":[],"_revision_number":1},{"id":"c926dfb64b570d88189b57321c3df479e1ac3ef2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-12 15:03:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9b08c403fb6942fa2387858d4aab64f7211ba483","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 23:48:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/54846483c9a94ed98f918f3c463ec87d : FAILURE in 4m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a334ea5eb53144c4a0c326319c794530 : SUCCESS in 4m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52f3c619397b408db9d86d4a55a02bfc : SUCCESS in 5m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f9b3d37239b74e3685286410d74ad7fc : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e66b51e48e94e909623dc5cfc2caf3b : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0b0c500154254b609447c8a6c7aa22de : SUCCESS in 6m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bccca3002c794feeb7813c37629787ac : SUCCESS in 5m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4631053ba8f04c9fb88e06dd2a04ee6e : SUCCESS in 11m 18s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/413f67d2d70449c3bd123fb2bed60f96 : SUCCESS in 12m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/381efc18ca6449be848313e3854d9fb6 : SUCCESS in 9m 56s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/0f084a7ed98841c5bce0df391f476ee5 : SUCCESS in 11m 29s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/75e423d1f3684604a8928d53d1299d9d : SUCCESS in 13m 31s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/882d5cee1fec441fb47f7b1c365a7c2d : SUCCESS in 7m 48s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/4f095451c0db4b429abb1af86cf4d819 : SUCCESS in 6m 59s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/62f77a569ef64c2e89fae4c8166f1d9a : SUCCESS in 14m 28s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/cc8eee60fb944dc1855c3f148da17837 : SUCCESS in 12m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/902ab826db0a4054815e18acca67e851 : ERROR Unable to find playbook /var/lib/zuul/builds/902ab826db0a4054815e18acca67e851/untrusted/project_0/opendev.org/zuul/zuul-jobs/playbooks/javascript/pre-test.yaml in 6s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/489da295a1d94562987eb70263864475 : SUCCESS in 25m 54s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/c711243e97b34cdd8a1cc0008bbbcea5 : SUCCESS in 27m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a98e1b5e52c74682a8fcab5ace67512b : SUCCESS in 21m 04s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/2d5f0dfeff2b41f481e6fd506e5163d2 : SUCCESS in 27m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c8515ebd081a4e05a42a2623179150da : SUCCESS in 14m 39s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/a626d061f0634130947c17ca404411d3 : SUCCESS in 16m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d8a488e078fe484cbd50d0f59b4581a3 : SUCCESS in 19m 21s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/36bd46366fba47a9bfa6e38d51ca14fd : SUCCESS in 6m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c4c25146f1da463a8394c9760f113d9b : SUCCESS in 9m 56s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/0a00cedb86564fdeb7829c00cd9d1e6f : SUCCESS in 9m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0056ea3d604b425490250477247288af : SUCCESS in 1h 14m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/15c20e24cf9f4e2ab5f7af996812213c : SUCCESS in 1h 40m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/acbf62b3e1a241a8a94cc9d00915c184 : SUCCESS in 57m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ca810ee0d2d5aaafc5f33cf7a53a048faac9d32e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-14 14:52:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8e9899607fb3bbc56d5aa377f81e9d1eed8b13ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 16:40:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a305d193dcdc42a58f859f8ceec7b97b : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/31c4bd38c5e24f8c9487e8a6a58d0083 : SUCCESS in 4m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/397f6ccda4134b69b73f5d536e8c6af7 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7fe6998acf94b93816770da58b6cda1 : SUCCESS in 7m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/15f7668acee6402b9e4a353d31aadd10 : SUCCESS in 4m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1e25bd464535448aab36fd7f98a02752 : SUCCESS in 4m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0c941a43478042c1962bdea64eebc419 : SUCCESS in 9m 00s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/df63d520dffe4be481c8b29084b69636 : SUCCESS in 16m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c81c84baf9ec4baa98ff4eae361c0e53 : SUCCESS in 9m 14s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/19d2f13f24cd481593b756be790b53f2 : SUCCESS in 10m 50s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/3419c06ced04452bab62c51c86d7cae8 : SUCCESS in 10m 47s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1930b11218e14d67b725e0ba09fa5da3 : SUCCESS in 8m 40s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/ed9c4c8203eb4d02af1726a9544e6665 : SUCCESS in 6m 37s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/52681be65c05446d8c7847229262ae30 : SUCCESS in 11m 43s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/d75c8bbbcc7243eab81abe29160f9a99 : SUCCESS in 14m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0218d06a3df84c1da7904c28ad17b3e4 : SUCCESS in 12m 54s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5d26503d482243aaa3200bf2de211166 : SUCCESS in 24m 24s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/4be2d195c1514a6ba5027b9d64bc7e1c : SUCCESS in 31m 53s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/383e6add50bd41169c83add331d8fdcd : SUCCESS in 27m 12s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/a4e0438f249245bfac788bb678649dcd : SUCCESS in 25m 47s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3fe2fbfabf8f4dc0b7be4a1be4a0bafd : SUCCESS in 13m 36s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/81fa1ea3e24b4a2281abbfcb98f4578c : SUCCESS in 14m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/30562e86c34b4d7aa244aef97c5a4419 : SUCCESS in 21m 15s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/969cda050aad4d21a00ea112e5f5fc59 : SUCCESS in 6m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ed87f3a3fc054b4e9079efe5fb84e8b4 : SUCCESS in 13m 12s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/b0219707e1214a768719dc2f6de737e0 : SUCCESS in 10m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f33cf1d65de74a42bef206fd1b237b60 : SUCCESS in 1h 02m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/213638e40bb04f87bff70610e7f03271 : SUCCESS in 1h 31m 40s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6bc34aa0b2264ee39e73cc5cdc477817 : SUCCESS in 1h 03m 31s","accounts_in_message":[],"_revision_number":1},{"id":"6d61a79513b661bd72a4021d8f5168107c1260ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 16:40:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1539c5f27e25b1b5280e420aec83b7b2f68d5701","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 18:35:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efb4569930b3456587c5d913ea5d09cd : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3df04a6822c4f4682875441586ca2b9 : SUCCESS in 5m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d390a714e0d14994b0447df6bc4488ed : SUCCESS in 9m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab9200198ce1499bbc7d4751868833eb : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/29ac1ca9cf7a48d089468dd31a0c1ec1 : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e366447df5645f4a7afc33fdd0d23a2 : SUCCESS in 3m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/aba881502e204d76b97d17d7509db3bd : SUCCESS in 8m 12s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/1171c6ae98804125a1d66115047a0ae7 : SUCCESS in 9m 07s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6145079f99f241ad8c0f4f22e8c3c167 : SUCCESS in 8m 11s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/d05af891a84443cc8a409ac0a3ab73b2 : SUCCESS in 12m 01s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/26a518d3cdeb4d44a7f575e2a5e25242 : SUCCESS in 5m 13s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/c01e35100e5f4c4b80131e3bb2181d5e : SUCCESS in 5m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a778572db4c14724a072d7ce6d040944 : SUCCESS in 12m 10s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/a100d98c0cf7481799d152c5da707b36 : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d91f3e0e6fca4597ac16bff277264fe7 : SUCCESS in 8m 11s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8445efbf6e724b6a94e76327cad0c88b : SUCCESS in 26m 55s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/5b1571e88f094eaa84b553048f75f178 : SUCCESS in 29m 55s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/90124317da954ef6b1ef2c146004459a : SUCCESS in 26m 42s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4eac0e2041bd42e381485ca8a40f5083 : SUCCESS in 26m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/cc0ab9ac03d34834907a1a927a4791cb : SUCCESS in 14m 24s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/66a47a6471634929b909960734807375 : SUCCESS in 12m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/99432f4ff39c4c07b2832085966a4c66 : SUCCESS in 20m 20s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/fba33dfb778942e9a7442a9c9ccbee22 : SUCCESS in 5m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a97b204934ba4cada5288588303c380b : SUCCESS in 11m 18s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/60f7d654b4424cca92138b9ecf813cd7 : SUCCESS in 9m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa8dea64be2e451ebe9ddb0d1c90f37f : SUCCESS in 1h 12m 46s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a551e9787f9447e5965b1e21da3a2867 : SUCCESS in 1h 48m 57s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1941935bc8fe46e09f17753bad786427 : SUCCESS in 51m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d221b8c27b6e686499a366a362797b211e1ddb77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 18:35:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36294b23237c2a1eb02201d993fe1e79e00c9894","revisions":{"36294b23237c2a1eb02201d993fe1e79e00c9894":{"kind":"REWORK","_number":1,"created":"2020-05-08 09:45:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/20/726320/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/726320/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/726320/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/726320/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/726320/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/726320/1"}}},"commit":{"parents":[{"commit":"881814daf02179eb47be4f3ce248560b2b8199fd","subject":"update constraint for horizon to new release 15.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/881814daf02179eb47be4f3ce248560b2b8199fd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-08 09:45:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-08 09:45:19.000000000","tz":0},"subject":"update constraint for oslo.context to new release 2.22.2","message":"update constraint for oslo.context to new release 2.22.2\n\nChange-Id: I0f4042e5bb17fde2c861a1e2c086d0d51d126588\nmeta:version: 2.22.2\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I4b65b61593d28074e71508e8786befaf00e53f6e\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36294b23237c2a1eb02201d993fe1e79e00c9894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36294b23237c2a1eb02201d993fe1e79e00c9894"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
