)]}'
{"id":"openstack%2Frequirements~726322","triplet_id":"openstack%2Frequirements~stable%2Fstein~I321433f6b3a74459c4756a0067c466cfd8c055fe","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I321433f6b3a74459c4756a0067c466cfd8c055fe","subject":"update constraint for oslotest to new release 3.7.2","status":"MERGED","created":"2020-05-08 09:46:17.000000000","updated":"2020-05-14 23:12:34.000000000","submitted":"2020-05-14 23:12:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726322-1589497954397-fedd4e4a","meta_rev_id":"d00603e0a28ac481c81e2ee81df54d52579aa92a","_number":726322,"virtual_id_number":726322,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-14 23:12:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-09 12:48:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-14 18:06:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-14 18:06:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-09 12:48:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-14 18:06:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-14 23:12:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fdb1d5910bfb6a54476160395a4fd39d07c949dd","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-08 09:46:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f5b67ec69d1a9f9b49758fb991f1e297bf2f085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 11:28:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/47ca4f64cf0c4f83b06f0fbda11ba62e : FAILURE in 4m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bd89cb9aa6e4258a4500b8c810e5860 : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8a5a7d566a2b41a6ae6b3d5248f3fbdf : SUCCESS in 7m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/30305f5ace2b46c88e6945f2c67f477b : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aeac406a8c6b4428a2017987634a3240 : SUCCESS in 6m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b63ff7f554348408382dccf23d4960f : SUCCESS in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9de72553c0b74fb1adfa1687701d3740 : SUCCESS in 6m 43s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/89efdc763c424dab8bf9dcca88a41b06 : SUCCESS in 9m 08s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/8e4bd1af007c41e48079d299b401aeb7 : SUCCESS in 9m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/085c1502bf4344428325bfc447aa286e : SUCCESS in 9m 11s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/7ae474ed5bad46668f71414092fcc8df : FAILURE in 6m 04s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/adf872c10cf3477cbad422598582ba3d : SUCCESS in 10m 12s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/fdc29d7eabc14129b61da0f1095f6804 : SUCCESS in 5m 31s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/2904449b4f7247ae93c5fc75ed95a8aa : SUCCESS in 5m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/988f6cf7ecbc4e3b8f41e65a18e3dc2c : SUCCESS in 10m 51s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/c06579f07bb5449082a2990edd228a5e : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83867d4be9c8402caebb242aeb87ee9a : SUCCESS in 9m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0bc97659167947fb90282beec1f8bc44 : SUCCESS in 22m 31s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/b86278960f0f41f2b90639bf08a7a24b : SUCCESS in 24m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/16cf50aca85f4e418c142ee4a7808ae1 : SUCCESS in 31m 15s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/c9735f4d23764a098d2177bd046b17ff : SUCCESS in 28m 18s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a056b792fa3641eea4c09547e93807b5 : FAILURE in 6m 59s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/1001bcada31645919ae0cbba93138c3e : SUCCESS in 13m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cada83d0ab4a44db8c70ac817552cc40 : FAILURE in 4m 47s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/f33d23196bdb46f6b5f9ec2223ad7cba : SUCCESS in 5m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6d2ab39305a14cb492c47d05e22d91f0 : SUCCESS in 9m 17s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/0958d70c271743f1a2f38aa90c9e564f : SUCCESS in 9m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/86c135c3458a4968a18c2b4edcc66ffb : SUCCESS in 1h 10m 29s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6e595293fdb64099a5b958b4da7b3e3e : SUCCESS in 1h 36m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bb427bbb6598427ca84d9bd8718eb60c : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":1},{"id":"eff7d5e7d855769afabd0a156b58d319805ab310","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-09 12:48:54.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck\n\npypi mirror update race","accounts_in_message":[],"_revision_number":1},{"id":"52ad45d6a6d41b2a9dd3684c88a1f74ae42616d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 14:20:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/f494f6980e334b82a8f89f3ea6d8f57c : FAILURE in 3m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d54cc4edefc9433b8223c6ecee25e65d : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa4da66c725b4b638e2a96e682043fb0 : SUCCESS in 5m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c8947b2982474e0e97223111007c6f2d : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fd6f6583381452db84bb1b2fa80fa3f : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ef9f971ddf2746659d9b7fa91354c739 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3d2567aa8ef141e18bc8993e3c579eae : SUCCESS in 5m 09s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/f3d4beada6c74ba7ab0047d318b66949 : SUCCESS in 7m 56s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/f63e5a521de74e61bac89fa4b2b367e4 : SUCCESS in 11m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8560292fef844470ab88452f75a92b02 : SUCCESS in 7m 21s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/38dcb6865b284a6ab45c3a58358b3871 : SUCCESS in 11m 52s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/a6f93cfd7d9547549a7395a2c1e1c777 : SUCCESS in 8m 48s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4f74721ed76a421a81d495f415b27035 : SUCCESS in 6m 15s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/6cd5273223ea4e4bb76b0d072591b3d7 : SUCCESS in 6m 11s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/022ecdfb51484012a86791b0e95f781f : SUCCESS in 9m 56s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/6ffec8582d044871b88580c51c1f18ed : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1ea921a1f7ff4d89abd5378f3ba52380 : SUCCESS in 9m 01s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5d7a3f61c0ff4856bc0722cf9f1f1733 : SUCCESS in 23m 43s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/0ce6d3a6fac1449680216bd7475a8a22 : SUCCESS in 23m 39s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a5bf62642f634a5eb369a485db850e02 : SUCCESS in 18m 36s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/fe515adaa43a41dbb26b6a3452243c4b : SUCCESS in 27m 08s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6ea03edd429f46e2861ffce3c1f4dc28 : SUCCESS in 13m 28s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/c4cb266f83e2446f8709217b596bbf5c : SUCCESS in 13m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bcee5c264d074c2baa738a5819960c9f : SUCCESS in 17m 20s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/f8dd06d52acc4e00a96e67ab07f15de1 : SUCCESS in 6m 19s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5f1d955aad8f40389523f82c2ca69fb7 : SUCCESS in 9m 14s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/23007a806a0c4511ab2d13afbef302c2 : FAILURE in 8m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a740181224c4012bcf2cdd6506ad123 : SUCCESS in 1h 04m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9d45d570ae634c7db997412f7de2b60b : SUCCESS in 1h 31m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e98768ac212d4c23a638dfc5159a784a : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":1},{"id":"260d8b04f732b7f1dbac8714301ed237c9c1d9e9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-14 16:06:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5280d1a2da699f9751b89ffbbfc588a91f07ffba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 17:43:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7538f60f0a23475eb496d3e5fc9cb498 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c697d60e903946cb8d71e4692e386276 : SUCCESS in 7m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d4dae494079140dc91ffc45da5cd2740 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdc936a1ca8547a6b0b14305631f9b0d : SUCCESS in 5m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ad1310981914478afef785fcb44ec47 : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8cbb8b5b3f6a4c828d3b6d989d358572 : SUCCESS in 6m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c27b6f7d22f648ad9ce1978d52f4e667 : SUCCESS in 13m 35s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/747f09c2fa7f4259b6e0ebef3c070ee8 : SUCCESS in 14m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5234dd4dec6449b3a691222523976f8b : SUCCESS in 9m 10s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1397c0776a6e4651955989b4bb959d53 : SUCCESS in 10m 00s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/57ebbaed7db7480983fc11aa526e5e36 : SUCCESS in 11m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/be7447fdcb3d4d048830ac0a183aa74f : SUCCESS in 7m 32s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/5784b90972e64838af7fe7e3a92c6112 : SUCCESS in 7m 28s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/72cc17642e8644a3ade17731ba63a54a : SUCCESS in 13m 48s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/17d26004921a4097889e963f6f3d63d2 : SUCCESS in 13m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cbef85c41e2e4c1a9ab67b9e9892a074 : SUCCESS in 8m 12s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/d8c60e8c26054d8ea80d999612344e80 : SUCCESS in 27m 51s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/4910ebc448aa42249519527162452871 : SUCCESS in 24m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/aaac2aec967140ec939aba75dd3c19df : SUCCESS in 20m 45s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/18d49cb779c34b28bcdc6045bc9b81a1 : SUCCESS in 26m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a0be8b91770c42858b03c134272ab8f0 : SUCCESS in 12m 58s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/f8bb2a3988444b1cbfc1872fa9ca4494 : SUCCESS in 16m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9bb46da4e0c84361ac24ceefecd216f9 : SUCCESS in 21m 56s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/3a71a1c97080493aa7df57a223e1e8df : SUCCESS in 7m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/df255bd50fb74d4f840abbb07d447c59 : SUCCESS in 10m 41s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/20fea1ebdc4840be8bb8cc5c6e7714e7 : SUCCESS in 10m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1fec8a9dc9c24ac8921162f3584a886e : SUCCESS in 1h 02m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7a452b5a308f4a1fb452e9da0a020c02 : SUCCESS in 1h 31m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/42c100f0000546cdb0c24ceda4d7146e : SUCCESS in 1h 06m 16s","accounts_in_message":[],"_revision_number":1},{"id":"51eec61901b61bcb9caef3fc7f5587b4224beefd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-14 18:06:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"273bb2cc403c372e45c0f31a0ddee912caff5255","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 18:06:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b8df60bf4efb11c132e6e6dbf30e1cd0201d703","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:48:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c01c6cfd24b448fa60ca191e351a7ad : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6102adccbe704541842d21edde08c9ee : SUCCESS in 5m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7e7ff7b62d904333bc28578cd216c361 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83a9103c3c994c59bbd46bcf1f449cd9 : SUCCESS in 7m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6f3de6838eb940e5b2de012f10aad1c3 : SUCCESS in 8m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d8745559a72f4f1ab67fe9b171639f59 : SUCCESS in 5m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4ce3dd8fd21b4c0eb190f8d2a2da1f2c : SUCCESS in 10m 28s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/7ee65f7e6ced46509a1befa1566924c9 : SUCCESS in 9m 36s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b21a1eef5f404dab95392f59b2554298 : SUCCESS in 13m 08s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/419aa5efca35476f9747f156a426ae9e : SUCCESS in 12m 12s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a3738c60632a41e885244bd1df09d15f : SUCCESS in 7m 23s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/437f383518ef42688171d3fa4a1c57e7 : SUCCESS in 7m 33s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b894f2da9dca4be0b44e37c457419f4b : SUCCESS in 12m 16s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/5ceb92dcc22a478caa4f4ec51a26de88 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f331acf09693459ea773b81b81af1b49 : SUCCESS in 9m 20s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/32265e9b4c504061a7a82b995c658b5c : SUCCESS in 33m 52s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/bd45fb7c508844e5b7f295a8335fdcec : TIMED_OUT in 43m 59s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/87db206cdb134a9f9a73370cd8ca3cdc : SUCCESS in 24m 46s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/ff3f6210075f449bad3b47c74b82e665 : SUCCESS in 29m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a9a43312260449b6ac98b78fd644b1a1 : SUCCESS in 14m 41s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/5f73ccfe680f449aa13a5da43cb0e098 : SUCCESS in 22m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bee199bb21dd4d4fac9777ea38ef7963 : SUCCESS in 20m 43s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/883ff47198904dc59f93eb7fda8c107f : SUCCESS in 5m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6738d1a1c21843d0b81f9c83369cc164 : SUCCESS in 8m 47s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/e5648e36581d414e8ed8801152026e97 : SUCCESS in 8m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/41f558d64edf487882f1b47c89ea56c2 : SUCCESS in 1h 15m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3a535ad375d4493a81fc8f17a6db667d : SUCCESS in 1h 35m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a29e62051beb4078bb8b1d8497d957b9 : SUCCESS in 59m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ac79c34b16babdb83e86b2d5b886c51128b75c6a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-14 19:53:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a7011ff516e3e7d68c76c067d9a382c3d514eafc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 21:42:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f716f3e14be4eaa827821ed791549f3 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09808248b870413eb6f6718b039a072f : SUCCESS in 6m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8b3a69f647da4402bf481beadc906b26 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6bf51258be145bb942ec6f07fe18580 : SUCCESS in 7m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9dd2fcd6a30545a0aee8059933a5a712 : SUCCESS in 5m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/90770e05dc1f4e6eaaecd35ce2593f7e : SUCCESS in 4m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/55af7d2241804d718e0aab760a03927e : SUCCESS in 13m 14s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/7342662ea4bf4eb98119044efa0513e8 : SUCCESS in 8m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eeb6462943944304884c573a34b31aed : SUCCESS in 9m 32s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a012bf4589624fbf8ea36ebed972d6f8 : SUCCESS in 12m 54s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/51a5a28769124002b653164a4d44ce02 : SUCCESS in 9m 28s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c9d9d9203fb44992a24ca1522d2cdd84 : SUCCESS in 6m 12s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/c28cfe12a5724ab89f083736395f6c05 : SUCCESS in 7m 25s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/b5e25f72584f47c59f495599ff26f367 : SUCCESS in 11m 01s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/ac4ec1158ff7422294321d9ad719b07d : SUCCESS in 12m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc8d4ae47b26457bb078b4dcc82c6623 : SUCCESS in 10m 38s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ce8bc38475574a71b0ee106fd7f1ad98 : SUCCESS in 27m 29s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/9d0eee0b6fc14b0da7ec21dc66c283f3 : SUCCESS in 31m 21s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7a26ed18f4d04c6bb7beea73943fa0c9 : SUCCESS in 26m 59s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/107946e382a54a5387473cb4f6ea1890 : SUCCESS in 24m 29s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/1c8fb171c17a400d8d39ef1d51aa52ff : SUCCESS in 22m 34s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/175d5ebbc9ea4e65a54af1527213e5f5 : SUCCESS in 23m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c662d79b3f174504bb55a00e5ca7cd09 : SUCCESS in 24m 28s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/56dfe746b68f41ee969da37d3fabebc1 : SUCCESS in 6m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/48bdd5cfdcec40fba4822b7ed26d66d9 : SUCCESS in 16m 01s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/36964437ee8849fd879d4d5738de41b0 : SUCCESS in 14m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/51d3316957bb46c197e6d52d453faa71 : SUCCESS in 1h 03m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/68bc6c19a4f949999d37ca2f0bfe1c09 : SUCCESS in 1h 46m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/600a418b73a444bb9aaf5da458ce2f03 : SUCCESS in 57m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0e6f58171491fc8f94202f26c4a4f5c65f177066","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 21:42:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48598754c1e57845255af1a0744a1bbf3036df2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 23:12:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"86e78539ee981cf10e1c849a6222ee8029c023b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 23:12:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b30f0dae96148afb361993e3623ed94 : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/854e6bf8093749da8d3ec30883864817 : SUCCESS in 4m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/83008d6a200a4abda0b3da009b9dcc73 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02de49e21c3141628b5bf8386b97de47 : SUCCESS in 3m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c17c1f09b19f489a9ac38f0d62a8ab2b : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a09b383d5db46768e465dc374278643 : SUCCESS in 4m 05s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/6eef8e7b8acc407385230219f4b44923 : SUCCESS in 7m 59s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/656e393f6dae456caec2e7f45853babb : SUCCESS in 7m 42s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/78605e9a5e7d4a1c8234c3624cb893a1 : SUCCESS in 8m 34s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/b206946c4cf648a68652f79e0d93e39d : SUCCESS in 10m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/fb621a0ff8074f9c8f246774f95f625d : SUCCESS in 5m 32s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/bc0a67dda67e4bf5b8b605845fcd1584 : SUCCESS in 5m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/af5d33396e08415c97e8fc82c9d21e7d : SUCCESS in 10m 24s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/57e2a663dd384177baa8cbd9346f5d5c : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f0eb9907fa4c454da49cc833ceb6cf5e : SUCCESS in 8m 00s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ab8ee203a0d44c798b123840f7b26513 : SUCCESS in 21m 00s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/6953b30cd1f548b989419d1af6c0b1ca : SUCCESS in 23m 58s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/00626f9de7594dd3b5336452924bd1f7 : SUCCESS in 28m 02s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/944021e9bf394b27b2f0079c419b0ae3 : SUCCESS in 29m 02s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/852607145bbd41f68c6d94fda8e51618 : SUCCESS in 13m 28s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/2d421a9d35b34c62a5ae0cc557024d7b : SUCCESS in 18m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/988266db22cb433293c5d5a3850c5e8f : SUCCESS in 24m 28s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/c8cec6d78172435e930f4e5137a7ab61 : SUCCESS in 4m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b3116d6519c64746a6774033004b4f03 : SUCCESS in 12m 24s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/4fcd7adac3bd470693ce0f14d0d0abd9 : SUCCESS in 11m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd9f1aaec22a424f8f4664202f284a38 : SUCCESS in 1h 02m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/326d347e843e4dbcb6dbd02112ed6abc : SUCCESS in 1h 24m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/31976e63d5c0416b9dcac3ef1d660738 : SUCCESS in 53m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b7490b4c40e99ad0a422d9033c37f467fc9b8fb","revisions":{"0b7490b4c40e99ad0a422d9033c37f467fc9b8fb":{"kind":"REWORK","_number":1,"created":"2020-05-08 09:46:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/726322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/726322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/726322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/726322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/726322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/726322/1"}}},"commit":{"parents":[{"commit":"881814daf02179eb47be4f3ce248560b2b8199fd","subject":"update constraint for horizon to new release 15.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/881814daf02179eb47be4f3ce248560b2b8199fd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-08 09:46:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-08 09:46:12.000000000","tz":0},"subject":"update constraint for oslotest to new release 3.7.2","message":"update constraint for oslotest to new release 3.7.2\n\nChange-Id: I321433f6b3a74459c4756a0067c466cfd8c055fe\nmeta:version: 3.7.2\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I4b65b61593d28074e71508e8786befaf00e53f6e\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b7490b4c40e99ad0a422d9033c37f467fc9b8fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b7490b4c40e99ad0a422d9033c37f467fc9b8fb"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
