)]}'
{"id":"openstack%2Frequirements~726326","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Id7e4f3bddab7654be341385c69a73c3fa98f61ef","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Id7e4f3bddab7654be341385c69a73c3fa98f61ef","subject":"update constraint for castellan to new release 1.3.2","status":"MERGED","created":"2020-05-08 09:48:03.000000000","updated":"2020-05-08 17:18:56.000000000","submitted":"2020-05-08 17:18:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726326-1588958336231-a3f20a08","meta_rev_id":"87838b877bfe559d679ff67116f33d19178a592c","_number":726326,"virtual_id_number":726326,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-08 17:18:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-05-08 15:34:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-05-08 15:34:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-08 15:34:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-08 17:18:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3a156ead75b7d80314697783b05d93095597e6b7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-08 09:48:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b392adf40dc9eb4df4d1e38803e069dad4518538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 11:49:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/128e8252b1ca4d1295724a21ed299be1 : FAILURE in 4m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fd7f362c1be45d786280dfdf8254606 : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e09e3777dd0343ac8e4646dd64c91c4e : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87b9751ddd3b40fa8a75da18d6b48195 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c3ec0ce4fd8e4725aa54e5ebe6a922e9 : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ddfb226de9140a39916fac826262984 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6d376c4403c34429b75b2096d5763268 : SUCCESS in 4m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/dd84660e1eb14c579195534aa0258f87 : SUCCESS in 9m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9e12c62f91174951afaabf759fb41272 : SUCCESS in 10m 18s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e802814ff9614169991fefb233148930 : SUCCESS in 9m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9f53b086dd464e93b99ad532acf41857 : SUCCESS in 10m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3817e458304240eeac19aade369e4d5e : SUCCESS in 6m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/13bbf30ee92642419fc592ac3e6c336c : SUCCESS in 5m 56s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/dfe56d82f45e42c2b2e172bc9022dd50 : SUCCESS in 12m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0ae82619c1c24269a9effb4399acb61c : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/03ee908b478f4efb91f046baed083452 : SUCCESS in 8m 00s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/cd7ac0c431214d86b3f9d1402b40903d : SUCCESS in 15m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a3d4480fe81e49568a2b3ac8ced5e0cc : SUCCESS in 13m 03s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/265439f0b9d64a1699955ac4fbac2d21 : SUCCESS in 25m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f306ad4e6e304530afdb156845903c26 : SUCCESS in 26m 10s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/7876984748e941e7985f8c13b646e855 : SUCCESS in 13m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a5892095bc9b47f1a93d2a7360673070 : SUCCESS in 13m 42s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/02e0353ed7f34dfaa2ff79563186facb : SUCCESS in 14m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/43d431b73c734cccb4619ef9db326a37 : SUCCESS in 18m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5d56f5a0bdeb43948441ef8cf8877be3 : SUCCESS in 5m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1b658762950945b9a9ad6dc41f94fa14 : SUCCESS in 10m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/abd577c15337437fa73c82feb8f7df27 : SUCCESS in 10m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/59e6548f71494f7a96627b6beeffd254 : SUCCESS in 1h 26m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/82cf3f302b65494584b03ee75eb0ab8a : SUCCESS in 1h 51m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/51a92b5fa2a24568a8f0fd79a66f7a09 : SUCCESS in 55m 43s","accounts_in_message":[],"_revision_number":1},{"id":"00d82b42271ef399e86bd98a63ea1185c505d5ba","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-08 15:34:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e7f2e8d332fa759b8023e66435c30505f6879a25","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-08 15:34:03.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0e171f3b2afeac8904b980b07a83bc07543399b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 15:34:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88806c67a1a3e3ff4fbfd99b8b86715fb9f8cc29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 17:18:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd5478a1d6440eba5325ae605136f4b : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fd95710b0fc4b96b43fd1a76e25c3c0 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63d411b539ef4d97afc299da5cad0a8d : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c16f82ae9dd458899e30235694770c6 : SUCCESS in 6m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a54796344b64adeb29b673447734476 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dd923cccc21046d09aa3de19f3fd3d7a : SUCCESS in 4m 49s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e7aad28bb81f4778b585e28e68e5b76c : SUCCESS in 8m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3376bd7e1cc44bbb8bf4c24f88a5cf2d : SUCCESS in 8m 17s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/62e61248469e4cd492309d8b86b72f71 : SUCCESS in 10m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c5b603b4f0ce4d1fbef946de1af6c964 : SUCCESS in 11m 10s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/92fe2d9a815540f686b4269aa896a372 : SUCCESS in 5m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8059077b85004787a29e63b9052c28fe : SUCCESS in 5m 01s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4e837b37feeb492b90f4941e9bf07e17 : SUCCESS in 11m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ab69136501a84e9dbf2ecdff96650498 : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/40789d6f09154e06a8611a274778101c : SUCCESS in 8m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1e37be9d0462469da32bf75089b11d57 : SUCCESS in 19m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/64cdccb781754e3d8908882073e44271 : SUCCESS in 32m 47s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b6e0dadd004d47588ea4a906a4dcec65 : SUCCESS in 22m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fc2204ca04ec4e55a24adb095cb345b9 : SUCCESS in 20m 03s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/574b98e95af0413890da296cfeeaecf5 : SUCCESS in 16m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/11f730e37ac047e78f63eb40f6e24838 : SUCCESS in 14m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/37aabce4a5c24ab292a2cafa656c53ac : SUCCESS in 23m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8bf2e624a03742b9a0b9f662c8dddec0 : SUCCESS in 18m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/05065cac61634718a59f42322d2f4452 : SUCCESS in 5m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a361b76ba5364ff989724920da890e47 : SUCCESS in 9m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3ec04628c9d8461c9376f32a5bac8242 : SUCCESS in 11m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d2e1b1b289c4fe4a3855ffde5c81c70 : SUCCESS in 1h 16m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1aa1af9349ce41039fd42aa7b0cd092a : SUCCESS in 1h 39m 14s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f311c651d97b432ab74a6fd06216ba9c : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e0f81ed1f52b15ae412023561d3bca89cd5070af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 17:18:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"311b2b41f693e656ee38ced0b9868fdf7c148b22","revisions":{"311b2b41f693e656ee38ced0b9868fdf7c148b22":{"kind":"REWORK","_number":1,"created":"2020-05-08 09:48:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/26/726326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/26/726326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/726326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/726326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/726326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/26/726326/1"}}},"commit":{"parents":[{"commit":"ef36a239ea199b9deb81ceb081068f980f23693b","subject":"update constraint for paunch to new release 5.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef36a239ea199b9deb81ceb081068f980f23693b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-08 09:48:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-08 09:48:02.000000000","tz":0},"subject":"update constraint for castellan to new release 1.3.2","message":"update constraint for castellan to new release 1.3.2\n\nChange-Id: Id7e4f3bddab7654be341385c69a73c3fa98f61ef\nmeta:version: 1.3.2\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: I85b5b12b1ed12bbb9e8b67e68211f6cd6e6dc658\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/311b2b41f693e656ee38ced0b9868fdf7c148b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/311b2b41f693e656ee38ced0b9868fdf7c148b22"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
