)]}'
{"id":"openstack%2Frequirements~726346","triplet_id":"openstack%2Frequirements~master~I71c6f072c0a0b8de05830e43a81795f84cd1ab24","project":"openstack/requirements","branch":"master","topic":"dnspython","hashtags":[],"change_id":"I71c6f072c0a0b8de05830e43a81795f84cd1ab24","subject":"Update dnspython to remove python_version cap","status":"ABANDONED","created":"2020-05-08 11:18:32.000000000","updated":"2020-05-08 21:36:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"51c3e334e70f4d9bd0ee18929e7f943ad23fe8bb","_number":726346,"virtual_id_number":726346,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-05-08 16:07:00.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-08 13:47:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-08 13:47:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e025d20f5b89c1f12c1b773bb70120525f6d1d2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-08 11:18:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aaf2c8f21e527691139bdb053d472d473f01ec85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 13:47:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3eccd3a0d8b4650b89a2da5589ae615 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25e68ecec4c646a981f8c2e976cfc4eb : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f94b792d35bb419f89f758a4dacec421 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ac3b453071c4f79a91830f68d074706 : SUCCESS in 6m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5461eca4f58c4c45b81ea20c36f15655 : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d12a289ef79c475cb0ea4775e0d35d0d : SUCCESS in 5m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/860d33a7bb6f4ae5831127f0d11dda61 : FAILURE in 13m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/89cf26488c3c4095a5891ec1b368dfc8 : SUCCESS in 17m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/776655c8a35c4b09904d7d9fd7031443 : SUCCESS in 5m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/819762f4001a4219a94e4a9bf152cbf4 : SUCCESS in 12m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f4204ee47e6f49c882a7a0f4674fe817 : SUCCESS in 9m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/eb7c3babd2534b15880f461c79859dfc : SUCCESS in 19m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d2275e2f08cd48a4acc9e52ba8d6afb5 : SUCCESS in 35m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/76f405e20a784f7ca06564fa707c97c2 : SUCCESS in 13m 28s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/15dcc5b1a9594afdb32c17909ede2cc9 : SUCCESS in 16m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a14039abd0cf426cbf1dffa624eb3db3 : SUCCESS in 22m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1592e0b54feb4463914663f8deb60595 : SUCCESS in 6m 08s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/d698c83df2b04e9cbeb7025626031a7b : SUCCESS in 6m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a48b5be8138649eea9a028e313e12b6a : SUCCESS in 7m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3e09ca4f116a4b148f0c2caf8ea74f72 : SUCCESS in 12m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b0cf3f239aad4a9d9ee6537be9cf8678 : SUCCESS in 15m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42d9b01ebe074de292f7535c74422191 : SUCCESS in 51m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/80198d0b4f9c4b76a98aad74742b09ad : SUCCESS in 1h 11m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/58393e3c9e0a4a0291c43e05396a85a6 : SUCCESS in 1h 30m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/d057246eefac4a36b33e8ec620a9ec6a : SUCCESS in 1h 07m 35s","accounts_in_message":[],"_revision_number":1},{"id":"fb31378daba85f3afe9b70b6b3153ea261291588","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-08 16:07:00.000000000","message":"Patch Set 1:\n\nSo I think we may have a problem here. We need to raise dnspython to allow the current consumers of dnspython3 to move to it. But the last version of dnspython3 has a requirement for the older version of dnspython.\n\nWe might need to temporarily break designate (the only consumer of this) by skipping ahead to the step of removing dnspython3 and raising dnspython. Then that could be quickly followed by patches for the two designate repos that have this to switch over to dnspython.","accounts_in_message":[],"_revision_number":1},{"id":"186274b093b2b577d4418b6a5d0f3a13003549ca","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-08 21:36:29.000000000","message":"Abandoned\n\nGoing to have to go with the big bang.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f5d4be903f548e9b5c1fc491c6ac04ef3d31c2e","revisions":{"0f5d4be903f548e9b5c1fc491c6ac04ef3d31c2e":{"kind":"REWORK","_number":1,"created":"2020-05-08 11:18:32.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/46/726346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/726346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/726346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/726346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/726346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/726346/1"}}},"commit":{"parents":[{"commit":"aeeaafe47d912e5ab2b6266b2e18d28321d31c36","subject":"Merge \"Block influxdb\u003d\u003d5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeeaafe47d912e5ab2b6266b2e18d28321d31c36"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 11:16:14.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 11:16:14.000000000","tz":-300},"subject":"Update dnspython to remove python_version cap","message":"Update dnspython to remove python_version cap\n\ndnspython now supports py27 through py37. This is the first step to get\nsupport the two users of dnspython3 moved back over to dnspython in\norder to get rid of the no longer needed dnspython3 package.\n\nChange-Id: I71c6f072c0a0b8de05830e43a81795f84cd1ab24\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f5d4be903f548e9b5c1fc491c6ac04ef3d31c2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f5d4be903f548e9b5c1fc491c6ac04ef3d31c2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
