)]}'
{"id":"openstack%2Frequirements~726359","triplet_id":"openstack%2Frequirements~master~I4efb25bc500fa153004609acebfbf55c3b2363a7","project":"openstack/requirements","branch":"master","topic":"monotonic","hashtags":[],"change_id":"I4efb25bc500fa153004609acebfbf55c3b2363a7","subject":"Drop monotonic package from global reqs","status":"ABANDONED","created":"2020-05-08 11:35:37.000000000","updated":"2021-03-19 15:41:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"01527b5cd7e08d37943bb10e2215ee7362183653","_number":726359,"virtual_id_number":726359,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"date":"2020-12-04 19:07:55.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-23 20:57:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"recommended":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"all":[{"value":-2,"date":"2021-03-07 01:17:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-09-24 07:34:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":1,"date":"2020-09-23 15:22:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-09-23 15:04:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-19 10:10:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-23 15:04:27.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-09-23 15:22:24.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-09-23 20:57:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-24 07:34:36.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"},{"updated":"2020-10-24 01:45:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a86dc72a157e3e0fbdec6b54f248c662c3d5f759","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-08 11:35:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c60780706b3409e3ce7016b79befede61dbdd4c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 13:56:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b794917467e4054a15124252fe0c0de : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5863874b0a5247bd8dcd97b50559ae58 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ade610d9f73d47c59a5e8487d56fbf3f : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c959bb343fb7439dbc686087fa787af4 : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/243de4b162ef4df68b2b0ec6300422a0 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b7c0df426084de89eda56f42cf17784 : SUCCESS in 7m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/165238f0ea7e4f198df0a08a0e041c65 : SUCCESS in 9m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/965ccc7c61d941bf8a007712ca472d35 : SUCCESS in 10m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bb67462049184e9ea9b13de2e0018766 : SUCCESS in 6m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6fc6ab2bfb1b49569b1236f9c9ec3926 : SUCCESS in 14m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3442b573824340a88c51ebace93ac7e6 : SUCCESS in 10m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d5f95479639c4bff970adc406187f734 : SUCCESS in 13m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/63ee5b2102054859a0c6b928dd8df83c : SUCCESS in 31m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dc6b3d0ff5de4a3daee26516294f9ed7 : SUCCESS in 15m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0cdcc5c961284a689f47d44373600812 : SUCCESS in 17m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0127001fc60f49d697bf8c03f7f4052c : SUCCESS in 19m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3d49083ce314458ca06a825d420c3b64 : SUCCESS in 5m 20s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f2ce3f1c66fe410c875b2880d05c1072 : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f6bda802fe1435faddcc5e914589e04 : SUCCESS in 7m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/40a7a0aa7d9b463ca8f87ff0cfa80901 : SUCCESS in 13m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/60962a7180804d0cb2e50f1b5376ea96 : SUCCESS in 13m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/560310895cca4b33a3395f4dc845c920 : SUCCESS in 1h 06m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c4778ab0fb042409372a233b858f30f : SUCCESS in 1h 24m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2711200d8314f38ade2d63780e56424 : SUCCESS in 1h 24m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/8d4d4dafc6a2483cbb50915e03842763 : SUCCESS in 1h 12m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3167a61f29415fa027fc83073a5ef87cb1af48e1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-08 15:34:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c929b91e449ee3b5e3fd434ed4379915cbd3950e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 00:12:03.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d4db7a8c2567b9301defd4a389cc52a50172009","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 06:24:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a64a1ccb72e6a90d06aa669f306c2fb205aca2cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 09:08:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26e09ca5418543c28325939c995cd062 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/313ddcbcb0ea4f5f8ed8f9f081ceb49f : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/370ec40dcbb8432f8993677c35774b75 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdcd8c560cde4d139ed3a3b5d3a3483d : SUCCESS in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f5a5957b6aa24a53bed358cc32bc069c : SUCCESS in 6m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b0181e763a354eb3ad0f6b417eff31fc : SUCCESS in 5m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f1d72764aea34046a23fe9cc126df015 : SUCCESS in 9m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/102fa35b27c34e4285b16be2bada607f : SUCCESS in 11m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6d484f1967e1486f8fe66e9815df904f : SUCCESS in 6m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9db22254ec8f49f4bcab48fdb95b7bbd : SUCCESS in 12m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/243d7fe5790b4cea850770e9d0f811f9 : SUCCESS in 10m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c83f6e810ac244c18ca6c63fd027785d : SUCCESS in 15m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/81662e74d4134f9bac6bc8acded4bd58 : SUCCESS in 23m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1b05c20e35e146fbb938d85f0dedb36a : SUCCESS in 13m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8cf65b44612644819db0b418cd89f541 : SUCCESS in 13m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/583ca8fbd2554daa870844d3c2c70b9c : SUCCESS in 17m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d38e7b9f71fd483d8226abe015bf52ee : SUCCESS in 7m 08s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/624d5a89b8a6467687ca2ce8b38f5459 : SUCCESS in 7m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ed5a60006154f4bb0799c34b7b41b24 : SUCCESS in 7m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f34ab2a03ab443a6917d122edf684ecd : SUCCESS in 10m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e76907607527401a948ba0fa17d85747 : SUCCESS in 14m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f726f88b295a40feb16e42e52bbe39cb : SUCCESS in 56m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/195b0fdc9c944175a2c92763d315f8d9 : SUCCESS in 1h 07m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76ed44e0c79c463aafbdab47a01e9554 : SUCCESS in 1h 26m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/327a0dfd31b4401db71e029bdc6312ce : SUCCESS in 1h 13m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b6cda86c93c9a1de009d1d2f4cee26356acb79cf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 15:34:30.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"ced713ec026871de8b10c6fbf5343ecf94e6e760","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 15:34:35.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee032cc35568b04e9ec992ec178e56e952238f65","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-12 15:36:48.000000000","message":"Patch Set 1:\n\nThe two dependent patches have not merged yet. Some question on the nova one, and no reviews yet on the masakari one.","accounts_in_message":[],"_revision_number":1},{"id":"c79787ffacca117fb26ef99d381a0d05aa2093ec","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-17 18:51:21.000000000","message":"Patch Set 1: Code-Review+1\n\nnote this will need rebasing (unmergable now)","accounts_in_message":[],"_revision_number":1},{"id":"5337ee066113ca853040560c639c5bc4877be7f2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-17 18:52:54.000000000","message":"Patch Set 1: -Workflow\n\nstill needs masakari anyway","accounts_in_message":[],"_revision_number":1},{"id":"f5bd903e2051e44561c1178595cec907f78577e6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-17 19:07:06.000000000","message":"Patch Set 1:\n\n\u003e still needs masakari anyway\n\nYep, just waiting for some movement there before I rebase this so I don\u0027t need to keep doing it. Crossing fingers that that will be soon.","accounts_in_message":[],"_revision_number":1},{"id":"e5fe33e890e1672c8432ac72c09109b5817c8ae3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-17 19:26:31.000000000","message":"Patch Set 1:\n\nI just noticed eventlet still pulls monotonic in, shame.","accounts_in_message":[],"_revision_number":1},{"id":"93236d2820b164acb4ce601eaf0e087e501dccd5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-17 19:33:11.000000000","message":"Patch Set 1:\n\nstill allows us to remove it from global-requirements","accounts_in_message":[],"_revision_number":1},{"id":"e4d9c422b485c336f3fd43fbec073aaadd782229","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-17 19:35:40.000000000","message":"Patch Set 1:\n\nYeah, I\u0027m just thinking, hypothetically, since e.g. Kolla depends on u-c blocking broken newer deps, it now allows new monotonic version to break. Just some food for thought. :-)","accounts_in_message":[],"_revision_number":1},{"id":"b32f2d040c47bc12f5ec74c75140077a8855d53a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-19 10:10:21.000000000","message":"Patch Set 1: Code-Review+1\n\nShould be rebased and conflict fixed","accounts_in_message":[],"_revision_number":1},{"id":"80a53aa3f7ef88cc4430e6b1f1d8f5b3a22adc35","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-26 17:00:34.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"053fb1072c23fc0b024e27de3dda48c4fbcef728","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-26 17:53:33.000000000","message":"Patch Set 1:\n\nmasakari in but this needs conflict fixing","accounts_in_message":[],"_revision_number":1},{"id":"0a0f9779a169babced7a017e0045700de900299d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 18:06:43.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"84d65cc3643e736785917322df7c922bffbbbc4d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-08-26 18:08:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0fd2fd058f3a941140ac9f650648a7304b11ed77","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-26 18:11:42.000000000","message":"Patch Set 2:\n\nhmm, if we are only removing it from upper-constraints it will be auto-removed if appropriate on the next bot run (it\u0027s already not in global-reqs).","accounts_in_message":[],"_revision_number":2},{"id":"9119a00a2611003fa22a265065e1452a47e5d9b6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-26 18:12:38.000000000","message":"Patch Set 2: Code-Review+1\n\ndoes not hurt to do it here","accounts_in_message":[],"_revision_number":2},{"id":"f5bfb50675369d71bcafd1131f3b9f76df925a2f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-08-26 18:16:22.000000000","message":"Patch Set 2: Code-Review-1\n\nIt might actually hurt if it\u0027s still needed as a dep of a dep (of a dep).  We may have removed it from openstack, but that does not mean that it\u0027s still not brought in a s dependency elsewhere.","accounts_in_message":[],"_revision_number":2},{"id":"539744fb89219766d02d334036c4ffb550abe999","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 19:15:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/17e504b6ceaa44bcbbc4fc74a4b87dae : SUCCESS in 1h 03m 05s","accounts_in_message":[],"_revision_number":2},{"id":"8b9f3b4a62aeeef778009bc3ad1868242a939d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 21:59:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ca5cfd1ec9a41b48986cad21c3d39bb : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f36f447e7a6445aad5c97c4cda4e720 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4233e3062868466f8838a8837cdea189 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edb45160f3d449c7a2c93d2ad05c1c8f : SUCCESS in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/57a7fcc0b49849ee85ec2b2c2aa7d5ea : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/185c8a5ee3f6462d86caa9b957200bf6 : SUCCESS in 6m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/58e0b2fadd614c6ba9a81f538aa533d2 : SUCCESS in 8m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/973dbeb8dc1d4695862be79644497e62 : SUCCESS in 12m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5cb103b186a842959eab1ee32aa2ff7c : SUCCESS in 8m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c93db5f7782843fca640911172532dbf : SUCCESS in 4m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a04983f92416434bb1cfce5ad00fd307 : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5a17c1775329480babf79ab68c5bcaae : SUCCESS in 6m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/53e72e1bfd814f34ad64b1e397a4f49d : SUCCESS in 7m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7b98339bdd5f4f55ae592f91c747b6d2 : SUCCESS in 23m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8eaf6f21649341f4b1fddd1841a2e887 : SUCCESS in 24m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7876f806a06f455da562b9da62e56ced : SUCCESS in 14m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9e0082bd0a2d4284a2348c2ec0c82085 : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be53fa3ab6df46b5bec6f3069c485f21 : SUCCESS in 17m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2ad8b19df8ed4bf0886d4d1c95d81643 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c098ae064624509b4f200b324b2a9af : SUCCESS in 7m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7169a568d523455dabaf027ab6259439 : SUCCESS in 10m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8aa052a53bfe4a429031d4e2e04caca0 : SUCCESS in 50m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6e67ed31f504cdba90b8a355b99720e : POST_FAILURE in 9m 35s","accounts_in_message":[],"_revision_number":2},{"id":"5381d692bb9d84da46a5380b731fddfbeae8a633","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-27 10:02:35.000000000","message":"Patch Set 2:\n\nwell, it\u0027s unlikely to get any other release than 1.5 nowadays","accounts_in_message":[],"_revision_number":2},{"id":"f7c4e1d3d9f6b9d9e5841c18c4748ff793a44890","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-09-11 15:33:33.000000000","message":"Patch Set 2: Workflow-1\n\n-W for reqs freeze","accounts_in_message":[],"_revision_number":2},{"id":"2fdbe812decac99357dfe0a55e0fcd7dba7728af","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-23 14:51:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b3a4cdb4d1135bc43c025f04d21d2a72e6365272","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-23 14:52:30.000000000","message":"Patch Set 3:\n\nEventlet still uses monotonic, but I wouldn\u0027t expect it to change in any way that we would need to enforce it from our upper-constraints.","accounts_in_message":[],"_revision_number":3},{"id":"3cd2c375a19e3df1a7c47ebcebfc77e5d2b55d63","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-09-23 15:04:27.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c0ce533ddd5f91b383df01f81f16fc74d8048482","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-09-23 15:22:24.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f65374223f16f55cf9ecc42872697d26a6b9f8fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 16:08:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d99435a3c9814891afcbd418072b27e8 : SUCCESS in 1h 10m 11s","accounts_in_message":[],"_revision_number":3},{"id":"847dc40e1df74c76f670a1ce0ed456cd720584e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 20:57:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90b9f4aaaab84b45b1e74791f25d8678 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a3266a27c524c83bda7c06c51dcbd5d : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/753568aeccc64b4a866a922ecb7c8071 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c24d86ca5f9438498c285ad4ff23c6e : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e0834cd2bf944debb2d34c066b518567 : SUCCESS in 7m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b4c329e48bd849fe8637c9e0752e062c : SUCCESS in 4m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/544834419a864a3ebecf4874f2502852 : SUCCESS in 9m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e3050c0b296340b088cc5c9605da0e28 : SUCCESS in 8m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e09ea39afc0c4d9597a28f52a4a00b30 : SUCCESS in 10m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/69b71ea1b8e748f5a3a9b83bffdcd493 : SUCCESS in 4m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4d3c4d8dfdf248f6ade6bbbb342aa1ae : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d2c9dcb7a274f42a9895ee388c98eef : SUCCESS in 7m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/046259ffbc1a46cfad93edfc2cc43baa : SUCCESS in 7m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6a0b6611883b412a8a55ca6aee80b215 : SUCCESS in 16m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/82161fccfb2447a39e6c05af4c0de24e : SUCCESS in 26m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8e82a0d667d6476bb95d2d39aae17176 : SUCCESS in 13m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ff4c5533014b42cdb254e690d076627d : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d390f4f77acd4c1ba360a273bbfa2c84 : SUCCESS in 16m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d5e31e29ceaf493eafe54a4c4a3feecb : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/615d4e392fd24286b265237c8bfa6e34 : SUCCESS in 7m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ed57f1c6aba641bcba0236ff5ff4170a : SUCCESS in 13m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d567e8462afb406c92e56c6b00651a28 : SUCCESS in 52m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/186e8ea4579546cebc044c05af1b36ef : SUCCESS in 1h 30m 31s","accounts_in_message":[],"_revision_number":3},{"id":"bde0b00320620d53d226618259cc00b88422bd3c","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-24 07:34:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f97a2c9883d90d86efc2d7381b1ac23b433d3ac7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-10-24 01:45:56.000000000","message":"Patch Set 3: Code-Review-1\n\nshouldn\u0027t this be auto-removed by the bot?  Right now it\u0027ll just be brought in by in the next bot run if we remove it.","accounts_in_message":[],"_revision_number":3},{"id":"e81691fbe3a38655e850c63f68af457a8af04797","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-04 19:07:55.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e shouldn\u0027t this be auto-removed by the bot?  Right now it\u0027ll just be brought in by in the next bot run if we remove it.\n\nDoes that bot do that? I didn\u0027t think we had clean up like that, only updates. There are other unused packages that we have had to manually remove in the past.","accounts_in_message":[],"_revision_number":3},{"id":"4e0149903c38d7d2113e1fd5f1a060f189d46dcc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-07 01:17:17.000000000","message":"Patch Set 3: Code-Review-2\n\nupper-constraints is generated in full every time the generate-constraints job is run.\n\nsomething is still pulling it in, remove it now and it\u0027ll just get readded.  This should probably be abandoned","accounts_in_message":[],"_revision_number":3},{"id":"c347aeb8a620797e98c2631a11925e2a63cfd550","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-03-13 01:39:02.000000000","message":"Patch Set 3:\n\nDNM for freeze period","accounts_in_message":[],"_revision_number":3},{"id":"01527b5cd7e08d37943bb10e2215ee7362183653","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-03-19 15:41:20.000000000","message":"Abandoned\n\nLooks like ceilometer added, though I don\u0027t think it should be used anymore.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9f8d7d2d6041eca896f8b79c0fbdc2ee7f419d9c","revisions":{"88ed944bdfd8b75ff729ad49c73f8ca786a7db46":{"kind":"REWORK","_number":1,"created":"2020-05-08 11:35:37.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/59/726359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/726359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/726359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/726359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/726359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/726359/1"}}},"commit":{"parents":[{"commit":"aeeaafe47d912e5ab2b6266b2e18d28321d31c36","subject":"Merge \"Block influxdb\u003d\u003d5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeeaafe47d912e5ab2b6266b2e18d28321d31c36"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 11:34:22.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 11:34:22.000000000","tz":-300},"subject":"Drop monotonic package from global reqs","message":"Drop monotonic package from global reqs\n\nMonotonic was only needed for py2 compatibility. Now that projects have\nbeen moved over to time.monotonic, we can drop this req.\n\nDepends-on: https://review.opendev.org/725425\nDepends-on: https://review.opendev.org/725420\n\nChange-Id: I4efb25bc500fa153004609acebfbf55c3b2363a7\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88ed944bdfd8b75ff729ad49c73f8ca786a7db46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/88ed944bdfd8b75ff729ad49c73f8ca786a7db46"}]},"branch":"refs/heads/master"},"05bcf1f6cd780ec6b8f470fd047ccccf857977b7":{"kind":"REWORK","_number":2,"created":"2020-08-26 18:08:06.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/59/726359/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/726359/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/726359/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/726359/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/726359/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/726359/2"}}},"commit":{"parents":[{"commit":"cc8489e1eb0f87de5fb17648a5802489d56e8aee","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc8489e1eb0f87de5fb17648a5802489d56e8aee"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 11:34:22.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-08-26 18:07:58.000000000","tz":-300},"subject":"Drop monotonic package from global reqs","message":"Drop monotonic package from global reqs\n\nMonotonic was only needed for py2 compatibility. Now that projects have\nbeen moved over to time.monotonic, we can drop this req.\n\nDepends-on: https://review.opendev.org/725425\nDepends-on: https://review.opendev.org/725420\n\nChange-Id: I4efb25bc500fa153004609acebfbf55c3b2363a7\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05bcf1f6cd780ec6b8f470fd047ccccf857977b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05bcf1f6cd780ec6b8f470fd047ccccf857977b7"}]},"branch":"refs/heads/master"},"9f8d7d2d6041eca896f8b79c0fbdc2ee7f419d9c":{"kind":"REWORK","_number":3,"created":"2020-09-23 14:51:49.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/59/726359/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/726359/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/726359/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/726359/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/726359/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/726359/3"}}},"commit":{"parents":[{"commit":"41d9437e24935d18d811e5640573187cde3aa2cf","subject":"Merge \"update constraint for tempest to new release 25.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41d9437e24935d18d811e5640573187cde3aa2cf"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 11:34:22.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-23 14:51:36.000000000","tz":-300},"subject":"Drop monotonic package from global reqs","message":"Drop monotonic package from global reqs\n\nMonotonic was only needed for py2 compatibility. Now that projects have\nbeen moved over to time.monotonic, we can drop this req.\n\nChange-Id: I4efb25bc500fa153004609acebfbf55c3b2363a7\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f8d7d2d6041eca896f8b79c0fbdc2ee7f419d9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f8d7d2d6041eca896f8b79c0fbdc2ee7f419d9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
