)]}'
{"id":"openstack%2Frequirements~726403","triplet_id":"openstack%2Frequirements~master~I140c3a5fb71a8fb82ccb696bade5be13bb9b0ba0","project":"openstack/requirements","branch":"master","topic":"health-report","hashtags":[],"change_id":"I140c3a5fb71a8fb82ccb696bade5be13bb9b0ba0","subject":"Add tool to generate a health report for all reqs","status":"MERGED","created":"2020-05-08 14:35:56.000000000","updated":"2020-05-12 11:04:28.000000000","submitted":"2020-05-12 11:01:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726403-1589281307845-299f2a5a","meta_rev_id":"938fd64c9fd7757fbeeb280c4cb28584bf4553cb","_number":726403,"virtual_id_number":726403,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 11:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-05-12 00:12:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-05-12 00:12:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 00:12:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-12 11:01:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0afa8012c9aad746b3016ec6ba9103ab643b4739","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-08 14:35:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c6457d590fd3725aff62cd7daa67e23596b0920","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 16:12:48.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a54ac274e0324be39fe04b8675f34538 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3f6e2bc8a6347b88a9c64b5ef4cd884 : FAILURE in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/753543b4a564433caf2fc0b4f22dc0d8 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd5eeea97e714954b2f35d282747bad6 : SUCCESS in 5m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d4efbe2e28c4452b9ba7ea498e24481 : SUCCESS in 7m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5778a1eec7d1417e9e63ab3b28711c77 : SUCCESS in 1h 16m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f67fb922a40549a39d82c9ead5bba67c : SUCCESS in 1h 00m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04c9ba7b914f4588b6c53fef4f2aa056 : SUCCESS in 1h 35m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/4dafabff23c347719d4eb5fee4817c16 : SUCCESS in 1h 00m 11s","accounts_in_message":[],"_revision_number":1},{"id":"74b50730c172c7d568b15e7be6b1b40d73e7f7ac","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-08 16:14:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"97a704596a00296a9e42463c47c5cc271a0c8a6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-08 17:46:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/063444dea8e34785bf321337f7d6d65a : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e11c119e5c2d45e58c390810be140c65 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6972dc9365724c89a0fa60cd3f3528ee : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bc42715d0a7470eacd409f0fa33f22d : SUCCESS in 5m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/364e592e964f4b6b8379a6e964d31624 : SUCCESS in 4m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7407ae6213ad46b1b9240ea468d1ce19 : SUCCESS in 50m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b720d104248049da8915db67ad69fef7 : SUCCESS in 1h 24m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4cd7e6d97e24401fa8b0c6df804cdc27 : SUCCESS in 1h 21m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/b6241b9368264793a512493d00a1740f : SUCCESS in 1h 02m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b7c2730bfe6edf747384d16ba11807bd11dc93dc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 00:12:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5346938e871ea0df61f76277d7daf74495a4d7b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 00:12:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b490b65486719805c2a19547e5cd705d7c35c61f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 01:52:34.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/404677647e0d4675ae3cb03d050fa177 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58b2dd9ba995453db31c8e7e410e92c9 : FAILURE in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f05deb4154d4234908b52a22d6442bc : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35ce0683217a4a8c89ba60fb42c06a51 : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b548c8a031c84c79b43f2b698ad3b5cf : SUCCESS in 4m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c105a3d3ec874fcc9d4fa27a2ba68654 : SUCCESS in 57m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed10d7c26a0941c4bc92fa42216e6da6 : SUCCESS in 1h 13m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a24cb87bc1e407ba6a16cdb81a38551 : SUCCESS in 1h 20m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/293a6d6fb1fd45148bc6549e9adffdd1 : SUCCESS in 1h 11m 57s\n\nWarning:\n  Comments left for invalid file openstack_requirements/tests/test_update.py\n  Comments left for invalid file openstack_requirements/cmds/generate.py","accounts_in_message":[],"_revision_number":2},{"id":"af9227ca7ca1757b250f80f590ec06c97c9b1009","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 01:56:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"06519fcdd961db4de7214995d9816bf9d45d3b77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 03:20:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/084b5f639997467ca2bc2e40a44efa5b : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f56052edc045debdfbf02b721d4512 : FAILURE in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46b7957f661d4cc19b15ec69a43a23bd : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/822eea4704714d6fa4b3e0931fa0e6a1 : SUCCESS in 4m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d9d47ba097945e686d44994941eedee : SUCCESS in 4m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec3aa51b179f4b7baeb39d27f5b2727c : SUCCESS in 55m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/70d49f73dd174718882c681611572a93 : SUCCESS in 1h 14m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d03deb3a2484aff8c8ed3b2dd794ddd : SUCCESS in 1h 22m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/1122c4b459e64f7c895507d0072c79ab : SUCCESS in 1h 03m 28s\n\nWarning:\n  Comments left for invalid file openstack_requirements/cmds/generate.py\n  Comments left for invalid file openstack_requirements/tests/test_update.py","accounts_in_message":[],"_revision_number":2},{"id":"1bb023cb7c38760e77fc875d22bfc3de224eb2ae","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 06:24:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2f05d2d7f4442a651b48440539eb7c43188d6b61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 08:56:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37727f9ed76042269916811fc3d34e6e : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bbbca60e57e4b8bb2ef91bb4528f6e7 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/009f6c2afe4b4a929c08fb8d23ff3e23 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef280e9287a84a988e5c21f710399458 : SUCCESS in 5m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c4cc0214766414bb2cf4a138cd7daf6 : SUCCESS in 5m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2154933ca5214fa3acbc543c69eb25f4 : SUCCESS in 58m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/55b5401a1e364910a10d460cea9da99e : SUCCESS in 1h 10m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65d5d9694ec6438a9db798a5f1d956fc : SUCCESS in 1h 18m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/d98fe4ff7d4d4d7393b3b1815894068c : SUCCESS in 1h 06m 14s","accounts_in_message":[],"_revision_number":2},{"id":"63a5322efbb2e7145a0e6c8c7fa7aa10eb44a92a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 08:57:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"44083528bd502a0a7ad863f3d1157fef528b360b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 11:01:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0caf7827c3774f70bdf3f5569c57f09e : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/251137933c7941eda8a4ff3a429a3b40 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24aab9ef33db4e049614a5857be779b5 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8b278193ceb4c4ca64df731cbc86e79 : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3964acd238564962a1659905f5976d32 : SUCCESS in 3m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5468fcb5bf98464bb18416ff2abee25a : SUCCESS in 57m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c29d5242f7ba49fe8326b625a72e0c8e : SUCCESS in 1h 16m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23151e833b664665aadf08e7e18d7e78 : SUCCESS in 1h 28m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/d317f634ab49480880185de9d2df618e : SUCCESS in 57m 23s","accounts_in_message":[],"_revision_number":2},{"id":"1b7ed4495ef4a6b98555af9b64375d5904a4b2ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 11:01:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4b09c938226cc59a11c734aa9e61f3ad39788235","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 11:04:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb3467459d454d16a68f30c13b94b4c9 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b57da6e88845863ced1d264ca92f54b6f44143b3","revisions":{"ce0fadeb050622df569ad3b5eeaa457a95e0a8f6":{"kind":"REWORK","_number":1,"created":"2020-05-08 14:35:56.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/03/726403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/726403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/726403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/726403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/726403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/726403/1"}}},"commit":{"parents":[{"commit":"aeeaafe47d912e5ab2b6266b2e18d28321d31c36","subject":"Merge \"Block influxdb\u003d\u003d5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeeaafe47d912e5ab2b6266b2e18d28321d31c36"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 14:33:57.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 14:33:57.000000000","tz":-300},"subject":"Add tool to generate a health report for all reqs","message":"Add tool to generate a health report for all reqs\n\nThis adds a tool script that loops through all of our requirements and\noutput basic information for each one along with any things we might be\nconcerned about with each one. For now it just checks a few things, but\nwe could extend this issue checking as we find other things to check on.\n\nAlso refactors the unused package tool to put some common things into a\nshared function library used by both scripts.\n\nChange-Id: I140c3a5fb71a8fb82ccb696bade5be13bb9b0ba0\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ce0fadeb050622df569ad3b5eeaa457a95e0a8f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ce0fadeb050622df569ad3b5eeaa457a95e0a8f6"}]},"branch":"refs/heads/master"},"b57da6e88845863ced1d264ca92f54b6f44143b3":{"kind":"REWORK","_number":2,"created":"2020-05-08 16:14:59.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/03/726403/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/726403/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/726403/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/726403/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/726403/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/726403/2"}}},"commit":{"parents":[{"commit":"aeeaafe47d912e5ab2b6266b2e18d28321d31c36","subject":"Merge \"Block influxdb\u003d\u003d5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeeaafe47d912e5ab2b6266b2e18d28321d31c36"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 14:33:57.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-08 16:14:54.000000000","tz":-300},"subject":"Add tool to generate a health report for all reqs","message":"Add tool to generate a health report for all reqs\n\nThis adds a tool script that loops through all of our requirements and\noutput basic information for each one along with any things we might be\nconcerned about with each one. For now it just checks a few things, but\nwe could extend this issue checking as we find other things to check on.\n\nAlso refactors the unused package tool to put some common things into a\nshared function library used by both scripts.\n\nChange-Id: I140c3a5fb71a8fb82ccb696bade5be13bb9b0ba0\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b57da6e88845863ced1d264ca92f54b6f44143b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b57da6e88845863ced1d264ca92f54b6f44143b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
