)]}'
{"id":"openstack%2Frequirements~726931","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Id0c2371c9ba37a3f19c2e0a3c121f99b54ac9ce8","project":"openstack/requirements","branch":"stable/train","hashtags":[],"change_id":"Id0c2371c9ba37a3f19c2e0a3c121f99b54ac9ce8","subject":"Fix on stestr requirements for non-py2","status":"ABANDONED","created":"2020-05-11 19:13:06.000000000","updated":"2020-05-28 16:49:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e5688a2952b36a98f5e2164f4e6126598026f7db","_number":726931,"virtual_id_number":726931,"owner":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-05-11 22:14:45.000000000","_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-11 21:30:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-2,"date":"2020-05-11 19:27:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 19:13:06.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-05-11 19:27:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-11 21:30:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46b3cc52387a5f6768e2d4eca082fa283043c486","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-11 19:13:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b3db2f3df6236ae2ca7fde597c7462183c82c64","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 19:19:32.000000000","message":"Patch Set 1: Code-Review-1\n\n2.5.1 was the max version in stable/train. There is no need to raise the cap in order to blacklist a version that was not used for that release.","accounts_in_message":[],"_revision_number":1},{"id":"6bfffc93ce7eb939ccb0c599d9cfde17232ed976","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-11 19:20:56.000000000","message":"Patch Set 1:\n\n\u003e 2.5.1 was the max version in stable/train. There is no need to\n \u003e raise the cap in order to blacklist a version that was not used for\n \u003e that release.\n\nYou\u0027re right Sean. Good catch! Let me update.","accounts_in_message":[],"_revision_number":1},{"id":"6d18bc5ab9b79764a6a9d7cc1730a1ff072e31f2","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-11 19:22:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"89c78b5185ca8a54633206aa3db7f261c6aaf8c2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 19:27:34.000000000","message":"Patch Set 2: Code-Review-2\n\nNot needed.","accounts_in_message":[],"_revision_number":2},{"id":"b0163136b5a74a8e9484615b1ef333cc23e42924","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-11 19:32:17.000000000","message":"Patch Set 2:\n\n\u003e Not needed.\n\nSean, should we just fix Ocata directly then? We can\u0027t merge anything to stable/pike right now because of stestr brokenness.","accounts_in_message":[],"_revision_number":2},{"id":"390030eab5d77af9b2fcfeb751d3639be8e63db0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 19:35:00.000000000","message":"Patch Set 2:\n\n\u003e \u003e Not needed.\n \u003e \n \u003e Sean, should we just fix Ocata directly then? We can\u0027t merge\n \u003e anything to stable/pike right now because of stestr brokenness.\n\nWhich repo is this? If it is broken by a newer stestr, then it must be misconfigured to not be using these requirement constraints anyway.","accounts_in_message":[],"_revision_number":2},{"id":"2756b66fde6f33b592fc2f5e762e3415541305cc","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-11 19:48:51.000000000","message":"Patch Set 2:\n\n\u003e \u003e \u003e Not needed.\n \u003e \u003e\n \u003e \u003e Sean, should we just fix Ocata directly then? We can\u0027t merge\n \u003e \u003e anything to stable/pike right now because of stestr brokenness.\n \u003e \n \u003e Which repo is this? If it is broken by a newer stestr, then it must\n \u003e be misconfigured to not be using these requirement constraints\n \u003e anyway.\n\nThe problem is summerized here https://review.opendev.org/#/c/718202/.\nWe have a few options.\n\n1. backport https://review.opendev.org/#/c/715942/ as suggest by #openstack-infra\n2. fix https://review.opendev.org/#/c/718202/\n3. globally blacklist stestr version 3.0.0\n\nPlease advice.","accounts_in_message":[],"_revision_number":2},{"id":"69790be168ac6ec5675287d645a78881a5fb90d9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 20:01:07.000000000","message":"Patch Set 2:\n\n\u003e \u003e \u003e \u003e Not needed.\n \u003e \u003e \u003e\n \u003e \u003e \u003e Sean, should we just fix Ocata directly then? We can\u0027t merge\n \u003e \u003e \u003e anything to stable/pike right now because of stestr brokenness.\n \u003e \u003e\n \u003e \u003e Which repo is this? If it is broken by a newer stestr, then it\n \u003e must\n \u003e \u003e be misconfigured to not be using these requirement constraints\n \u003e \u003e anyway.\n \u003e \n \u003e The problem is summerized here https://review.opendev.org/#/c/718202/.\n \u003e We have a few options.\n \u003e \n \u003e 1. backport https://review.opendev.org/#/c/715942/ as suggest by\n \u003e #openstack-infra\n \u003e 2. fix https://review.opendev.org/#/c/718202/\n \u003e 3. globally blacklist stestr version 3.0.0\n \u003e \n \u003e Please advice.\n\nNo need to blacklist 3.0.0 because 3.0.0 is higher than the maximum allowed version already.\n\nHere is what caps for stable branches are:\nstable/train - 2.5.1\nstable/stein - 2.2.0\nstable/rocky - 2.1.0\nstable/queens - 2.2.0\nstable/pike - 2.0.0\nstable/ocata - not capped\n\nSo if anything, we need to add to stable/ocata. But I don\u0027t believe stestr was an official global requirement then, so I would be surprised.","accounts_in_message":[],"_revision_number":2},{"id":"d5596f2cef43d666b0f557b011bed4276416179f","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-11 20:12:12.000000000","message":"Patch Set 2:\n\n\u003e \u003e \u003e \u003e \u003e Not needed.\n \u003e \u003e \u003e \u003e\n \u003e \u003e \u003e \u003e Sean, should we just fix Ocata directly then? We can\u0027t merge\n \u003e \u003e \u003e \u003e anything to stable/pike right now because of stestr\n \u003e brokenness.\n \u003e \u003e \u003e\n \u003e \u003e \u003e Which repo is this? If it is broken by a newer stestr, then it\n \u003e \u003e must\n \u003e \u003e \u003e be misconfigured to not be using these requirement constraints\n \u003e \u003e \u003e anyway.\n \u003e \u003e\n \u003e \u003e The problem is summerized here https://review.opendev.org/#/c/718202/.\n \u003e \u003e We have a few options.\n \u003e \u003e\n \u003e \u003e 1. backport https://review.opendev.org/#/c/715942/ as suggest by\n \u003e \u003e #openstack-infra\n \u003e \u003e 2. fix https://review.opendev.org/#/c/718202/\n \u003e \u003e 3. globally blacklist stestr version 3.0.0\n \u003e \u003e\n \u003e \u003e Please advice.\n \u003e \n \u003e No need to blacklist 3.0.0 because 3.0.0 is higher than the maximum\n \u003e allowed version already.\n \u003e \n \u003e Here is what caps for stable branches are:\n \u003e stable/train - 2.5.1\n \u003e stable/stein - 2.2.0\n \u003e stable/rocky - 2.1.0\n \u003e stable/queens - 2.2.0\n \u003e stable/pike - 2.0.0\n \u003e stable/ocata - not capped\n \u003e \n \u003e So if anything, we need to add to stable/ocata. But I don\u0027t believe\n \u003e stestr was an official global requirement then, so I would be\n \u003e surprised.\n\nI agree fixing stable/ocata is the way to go. But since it was not an official global requirement and we can\u0027t be pulling in 3.0.0 either, blacklisting it seem to make more sense. This will afford packagers/distributors the flexibility to pick a version that their CI is comfortable with. What do you think?","accounts_in_message":[],"_revision_number":2},{"id":"d1ef2b87978d700d792dc4f5b67422200be29191","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-11 20:54:51.000000000","message":"Patch Set 2:\n\n\u003e \u003e Here is what caps for stable branches are:\n \u003e \u003e stable/train - 2.5.1\n \u003e \u003e stable/stein - 2.2.0\n \u003e \u003e stable/rocky - 2.1.0\n \u003e \u003e stable/queens - 2.2.0\n \u003e \u003e stable/pike - 2.0.0\n \u003e \u003e stable/ocata - not capped\n \u003e \u003e\n \u003e \u003e So if anything, we need to add to stable/ocata. But I don\u0027t\n \u003e believe\n \u003e \u003e stestr was an official global requirement then, so I would be\n \u003e \u003e surprised.\n \u003e \n \u003e I agree fixing stable/ocata is the way to go. But since it was not\n \u003e an official global requirement and we can\u0027t be pulling in 3.0.0\n \u003e either, blacklisting it seem to make more sense. This will afford\n \u003e packagers/distributors the flexibility to pick a version that their\n \u003e CI is comfortable with. What do you think?\n\nThere should be no need to blacklist it. If stestr were to be added to ocata, it should be the same version or less than the pike constraints. So at a minimum, it would have to be \u003c\u003d 2.0.0.","accounts_in_message":[],"_revision_number":2},{"id":"9aee6933f1ba94fdcc21f30d6afea897fc1a57c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 21:30:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/b89d621651984961bb40f48e0b5f2492 : FAILURE in 4m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/365c0b04bb4a4d958f5082dc19d6bcb8 : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86042ee89ea0424bb180e9a835ca561a : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b02c61b13b634324a11774fd95b5af8c : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4915a1ba05b42da93f8f834ba7cb8ee : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/02e9982201ed417394ec29e85d6acd0c : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/33ab452030db49cabb9709534d793628 : SUCCESS in 6m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/ba74ef9ef8d54bc380d17914b531def5 : SUCCESS in 9m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/df112e0f4a5a4eb7adbeec01765596d5 : SUCCESS in 9m 38s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/aff5cd2c1147438383cb7089bb1b8c71 : SUCCESS in 11m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dc0238fe3f72443c9869bc198ee495d5 : SUCCESS in 10m 02s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d63c6aa732e44bcfa86ebe812f2c2bf9 : SUCCESS in 6m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f3ba00a2a5d34807a6205aa271d176ce : SUCCESS in 6m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c81159b06238416583f432b2912ed6d2 : SUCCESS in 10m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/94ffdf63bebb42cfb1f543117f2863e6 : SUCCESS in 8m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/39b014ac0c9c48d39a4ab0a43bc45704 : SUCCESS in 7m 55s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/29f49d8601024a90940d2019eba2547e : SUCCESS in 13m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c9fddee30cfb4513930b37531139164b : SUCCESS in 14m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/37c5428b1a4746d0a3e553488b61d166 : SUCCESS in 25m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/44efe876a84245a99fbe14adc4f66b40 : SUCCESS in 24m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5005430ed67e43d88ef9163b11c100cd : SUCCESS in 21m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0efa2a27af2c4acd902ee45c9739f29a : SUCCESS in 22m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/59fbc1166f534cc8a264a93e106c75d7 : SUCCESS in 20m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b348f71e5fa4b40946e36564f4956eb : SUCCESS in 18m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/172f696e2d964b3cb1e43eaa2f10cfe3 : SUCCESS in 9m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3ca1fd31f7d244939b0b5e2f5a92b946 : SUCCESS in 12m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/12b4158583f84deabf37a357ec6aff18 : SUCCESS in 10m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/35286c1c1af646a98325d5138b67db38 : SUCCESS in 1h 04m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4c8624756c60423480f31ae1b3a9299e : SUCCESS in 1h 32m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/272f63c8ce544dadb740d62445777135 : SUCCESS in 55m 01s","accounts_in_message":[],"_revision_number":2},{"id":"c00bbb574f58fc651aba3643d9795bdb4477e36c","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-11 22:14:45.000000000","message":"Patch Set 2:\n\n\u003e \u003e \u003e Here is what caps for stable branches are:\n \u003e \u003e \u003e stable/train - 2.5.1\n \u003e \u003e \u003e stable/stein - 2.2.0\n \u003e \u003e \u003e stable/rocky - 2.1.0\n \u003e \u003e \u003e stable/queens - 2.2.0\n \u003e \u003e \u003e stable/pike - 2.0.0\n \u003e \u003e \u003e stable/ocata - not capped\n \u003e \u003e \u003e\n \u003e \u003e \u003e So if anything, we need to add to stable/ocata. But I don\u0027t\n \u003e \u003e believe\n \u003e \u003e \u003e stestr was an official global requirement then, so I would be\n \u003e \u003e \u003e surprised.\n \u003e \u003e\n \u003e \u003e I agree fixing stable/ocata is the way to go. But since it was\n \u003e not\n \u003e \u003e an official global requirement and we can\u0027t be pulling in 3.0.0\n \u003e \u003e either, blacklisting it seem to make more sense. This will afford\n \u003e \u003e packagers/distributors the flexibility to pick a version that\n \u003e their\n \u003e \u003e CI is comfortable with. What do you think?\n \u003e \n \u003e There should be no need to blacklist it. If stestr were to be added\n \u003e to ocata, it should be the same version or less than the pike\n \u003e constraints. So at a minimum, it would have to be \u003c\u003d 2.0.0.\n\nSounds good. I\u0027ll work off this one then https://review.opendev.org/#/c/718202/","accounts_in_message":[],"_revision_number":2},{"id":"e5688a2952b36a98f5e2164f4e6126598026f7db","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-05-28 16:49:36.000000000","message":"Abandoned\n\nNot needed","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3cea21e9895f6ad4fe81d4a401e3e0ae648442c2","revisions":{"bd8b07f5f330aa1814bd10636c780374d11e8d24":{"kind":"REWORK","_number":1,"created":"2020-05-11 19:13:06.000000000","uploader":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"ref":"refs/changes/31/726931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/726931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/726931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/726931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/726931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/726931/1"}}},"commit":{"parents":[{"commit":"311b2b41f693e656ee38ced0b9868fdf7c148b22","subject":"update constraint for castellan to new release 1.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/311b2b41f693e656ee38ced0b9868fdf7c148b22"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-30 13:31:48.000000000","tz":-300},"committer":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2020-05-11 19:12:36.000000000","tz":-420},"subject":"Fix on stestr requirements for non-py2","message":"Fix on stestr requirements for non-py2\n\nstestr 3.0.0 is released with drop the support of\npy2 but had some issue on setup.cfg metadata to restrict\nthe installation on \u003cpy3 env.\n- https://github.com/mtreinish/stestr/issues/292\n\nThis is fixed and released in 3.0.1\n\nLet\u0027s cap the stestr version for py2 and py3 env\nalso adding 3.0.0 in blacklist.\n\nChange-Id: Id0c2371c9ba37a3f19c2e0a3c121f99b54ac9ce8\n(cherry picked from commit ba693790ea2e85d4733b516d53813f62f890c0e1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd8b07f5f330aa1814bd10636c780374d11e8d24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd8b07f5f330aa1814bd10636c780374d11e8d24"}]},"branch":"refs/heads/stable/train"},"3cea21e9895f6ad4fe81d4a401e3e0ae648442c2":{"kind":"REWORK","_number":2,"created":"2020-05-11 19:22:52.000000000","uploader":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"ref":"refs/changes/31/726931/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/726931/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/726931/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/726931/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/726931/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/726931/2"}}},"commit":{"parents":[{"commit":"311b2b41f693e656ee38ced0b9868fdf7c148b22","subject":"update constraint for castellan to new release 1.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/311b2b41f693e656ee38ced0b9868fdf7c148b22"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-30 13:31:48.000000000","tz":-300},"committer":{"name":"Guang Yee","email":"guang.yee@suse.com","date":"2020-05-11 19:22:11.000000000","tz":-420},"subject":"Fix on stestr requirements for non-py2","message":"Fix on stestr requirements for non-py2\n\nstestr 3.0.0 is released with drop the support of\npy2 but had some issue on setup.cfg metadata to restrict\nthe installation on \u003cpy3 env.\n- https://github.com/mtreinish/stestr/issues/292\n\nThis is fixed and released in 3.0.1\n\nLet\u0027s cap the stestr version for py2 and py3 env\nalso adding 3.0.0 in blacklist.\n\nChange-Id: Id0c2371c9ba37a3f19c2e0a3c121f99b54ac9ce8\n(cherry picked from commit ba693790ea2e85d4733b516d53813f62f890c0e1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3cea21e9895f6ad4fe81d4a401e3e0ae648442c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3cea21e9895f6ad4fe81d4a401e3e0ae648442c2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
