)]}'
{"id":"openstack%2Frequirements~726982","triplet_id":"openstack%2Frequirements~stable%2Fpike~I29abcba929995f9307c392110160d889a29b47c3","project":"openstack/requirements","branch":"stable/pike","hashtags":[],"change_id":"I29abcba929995f9307c392110160d889a29b47c3","subject":"Fix job issues for stable branch","status":"MERGED","created":"2020-05-12 00:10:05.000000000","updated":"2020-05-13 05:54:40.000000000","submitted":"2020-05-13 05:54:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726982-1589349280601-838f9c63","meta_rev_id":"6e16dc1575bfb9b337d0f8906320f1ff47d9c119","_number":726982,"virtual_id_number":726982,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-13 05:54:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-05-12 15:44:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-05-12 15:44:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 15:44:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-13 05:54:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"646bf4e7643a0d0bee540e0a3bbee707857c31c7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-12 00:10:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d271606300f226639edf82604aa13d5d9aee4d8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 00:20:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1202cb6cda791ef22b4ba6186c0a8a29b15a1918","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 01:34:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/6717ea92c6c047608a23080a7417b692 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3d9500ca67436493e54561f66f94a3 : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/60e412564855415496b064ef51f96c04 : SUCCESS in 6m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e39c5d308ab24dd2beed45480a019573 : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f3ee41a8698646db8a284c96646f89b0 : SUCCESS in 4m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/775d2728b75f4262af2f551e6ed4e441 : FAILURE in 42m 47s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/60a09bc01ea842749ae358a2f37c8fcf : SUCCESS in 1h 22m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f889bedc61ba4c55bfa5f4e52508d4fd : FAILURE in 30m 18s","accounts_in_message":[],"_revision_number":1},{"id":"3f82b178044d224dde8ad0e1b6410d4b33dd96a5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-12 09:03:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c3b02c79626ef12b88a7df88095a447603b72e29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 11:44:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/8356e0575879459d99fcf6114e03a54f : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31057e0102c647cb804d721bb7c8cd78 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a450a7fe0a07423f954c075aae3f8185 : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b1d8b6c748a245fe9ec15e20ee95b69b : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9e23a106d6724c73a09d78a41fd9f283 : SUCCESS in 4m 20s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/6fcae32131d74e13bd2381c9eab071a0 : SUCCESS in 1h 26m 25s","accounts_in_message":[],"_revision_number":2},{"id":"665e5d75cac065963fd06792eabe60f09f3ee344","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 15:44:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"848b89e27b31ad8e11716b104b71bceb646e2497","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 15:45:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0ac08329ad1121c4bf13e61742a95f48d19c73f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 17:21:26.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b8996032e954595b1b2b5092513b5aa : SUCCESS in 8m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0fbfc5ace3404412a197b30c9a75caea : SUCCESS in 7m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/61e654b0c3834a85b0127f0399609a0c : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e669a1d9a5174f5e9e0120714482267e : SUCCESS in 5m 31s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/e26c12b30b384925bf1f5d2a5a182642 : FAILURE in 1h 28m 29s","accounts_in_message":[],"_revision_number":2},{"id":"c382aed244b50123f5c316a54d29d08102ce889a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 21:03:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"57f2c4399cf9d0613f264e3374c4823269b9a96b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 03:21:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7444a21ef4f34bc7b4103782c3e807ee : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c6d4a44ac034e5e812dd1c1857033f8 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c8b27ac0ea542e88cc085a1f4882fe5 : SUCCESS in 5m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7d8cfaa2b11c46c48a9b3ac82d5c50ff : SUCCESS in 5m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f99f273035c4f5b9cc21237f1bfca72 : SUCCESS in 4m 09s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/e0b0305cef9247dfbff8966b04b027c0 : SUCCESS in 1h 28m 39s","accounts_in_message":[],"_revision_number":2},{"id":"385eaae051a31915e749c9eee3ceb049f3756975","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 03:21:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"af0ec80db3722dcea7d57c03fc9b885f1946ce18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 05:54:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35390e5cd88d4efd9a842aacaec8e65e : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7173b2fe2bad418a86976eb3b2ffefb0 : SUCCESS in 4m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b5d652a1e38f4f45aa6fa614b260c613 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d28fa6d7910435a8b0288e881626e36 : SUCCESS in 3m 46s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/d2d0c4856b344c55a8aadeeb0c234ac0 : SUCCESS in 1h 33m 45s","accounts_in_message":[],"_revision_number":2},{"id":"6e16dc1575bfb9b337d0f8906320f1ff47d9c119","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 05:54:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0b82bf29b1c88155ea1d315cf6f694dbd408b42b","revisions":{"bd36221f1b242d5f2296ebe1ec23b58d6d84d085":{"kind":"REWORK","_number":1,"created":"2020-05-12 00:10:05.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/82/726982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/726982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/726982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/726982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/726982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/726982/1"}}},"commit":{"parents":[{"commit":"36c442553e29181c69fcd730c3cb145e3822adb0","subject":"demarcate python-memcached requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36c442553e29181c69fcd730c3cb145e3822adb0"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-11-12 16:56:02.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-12 00:09:32.000000000","tz":-300},"subject":"Remove requirements-integration job","message":"Remove requirements-integration job\n\nThis job was a legacy job that attempted to verify each project could be\nindependently installed after all global-requirements had been\ninstalled.\n\nThis job is no longer necessary since we have much better coverage with\nrequirements-check and other jobs that make sure the requirements are\ncompatible. It is also a legacy zuulv2 job, uses the long deprecated\ndevstack-gate, and runs with Python 2.7.\n\nFor historical reference, this was added originally to PBR back in 2013\n[0] and moved to requirements in 2014 [1].\n\n[0] https://opendev.org/openstack/pbr/commit/378261a8cba818a29b7e3d33b917de5ed0751579\n[1] https://opendev.org/openstack/requirements/commit/8e073738db06086f949d2be377999356aee2d8d0\n\nChange-Id: I29abcba929995f9307c392110160d889a29b47c3\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 58d622019d8e92b301699787180aafe16cde53ee)\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n(cherry picked from commit 442891e31e28bbc430be722ba50058ceda9e2308)\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n(cherry picked from commit 22b4b1d479bc046e0d9c4a38e2f3e2ef3d2c211a)\n(cherry picked from commit 337290363687289f139da62bf5f496e489e47c28)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd36221f1b242d5f2296ebe1ec23b58d6d84d085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd36221f1b242d5f2296ebe1ec23b58d6d84d085"}]},"branch":"refs/heads/stable/pike"},"0b82bf29b1c88155ea1d315cf6f694dbd408b42b":{"kind":"REWORK","_number":2,"created":"2020-05-12 09:03:02.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/82/726982/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/726982/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/726982/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/726982/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/726982/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/726982/2"}}},"commit":{"parents":[{"commit":"36c442553e29181c69fcd730c3cb145e3822adb0","subject":"demarcate python-memcached requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36c442553e29181c69fcd730c3cb145e3822adb0"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-12 08:57:58.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-12 09:01:39.000000000","tz":-300},"subject":"Fix job issues for stable branch","message":"Fix job issues for stable branch\n\nThis includes a few fixes that will need to merge together in order to\npass.\n\nStop testing grenade\n\nGrenade testing has a dependency on a newer stestr than what was\navailable in pike. Rather than rewrite the grenade job, since this\nbranch is in Extended Maintenance anyway, just drop the job.\n\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n\nfix pep8 errors\n\n(cherry picked from commit d92bd96ee1f20e829610511d24131f1888ddf37e)\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n\nRemove requirements-integration job\n\nThis job was a legacy job that attempted to verify each project could be\nindependently installed after all global-requirements had been\ninstalled.\n\nThis job is no longer necessary since we have much better coverage with\nrequirements-check and other jobs that make sure the requirements are\ncompatible. It is also a legacy zuulv2 job, uses the long deprecated\ndevstack-gate, and runs with Python 2.7.\n\nFor historical reference, this was added originally to PBR back in 2013\n[0] and moved to requirements in 2014 [1].\n\n[0] https://opendev.org/openstack/pbr/commit/378261a8cba818a29b7e3d33b917de5ed0751579\n[1] https://opendev.org/openstack/requirements/commit/8e073738db06086f949d2be377999356aee2d8d0\n\nChange-Id: I29abcba929995f9307c392110160d889a29b47c3\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 58d622019d8e92b301699787180aafe16cde53ee)\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n(cherry picked from commit 442891e31e28bbc430be722ba50058ceda9e2308)\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n(cherry picked from commit 22b4b1d479bc046e0d9c4a38e2f3e2ef3d2c211a)\n(cherry picked from commit 337290363687289f139da62bf5f496e489e47c28)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b82bf29b1c88155ea1d315cf6f694dbd408b42b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0b82bf29b1c88155ea1d315cf6f694dbd408b42b"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
