)]}'
{"id":"openstack%2Frequirements~726985","triplet_id":"openstack%2Frequirements~stable%2Fstein~I18928e1a7d4413423be36eeb87bbf6376ffbe2d9","project":"openstack/requirements","branch":"stable/stein","hashtags":[],"change_id":"I18928e1a7d4413423be36eeb87bbf6376ffbe2d9","subject":"Stop testing with pypy","status":"MERGED","created":"2020-05-12 00:14:38.000000000","updated":"2020-05-13 06:53:05.000000000","submitted":"2020-05-13 06:53:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726985-1589352785191-28ca5a1d","meta_rev_id":"f19a94c8856fb3962e92c0579304017cd6c2bfb9","_number":726985,"virtual_id_number":726985,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-13 06:53:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-05-12 00:15:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-05-12 00:15:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 00:15:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-13 06:53:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"162056f1a535296db4645bb936c702b8890246e3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-12 00:14:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c8e2a58f19c5d88631f22cada7822297ab089b6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 00:15:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d625b853dd92932ba6d4e6b6547ac6f2802f795e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 02:04:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ba382add2144960931a79c16a23adf8 : FAILURE in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ef641de38824313a3482ba3439d24c3 : SUCCESS in 4m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0bb3b8cb53464288a94b73f0c98b067e : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1388f4bab0d64e57a5f26774cb126410 : SUCCESS in 5m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/642285416ad741e19e2898ba579f8e6c : SUCCESS in 3m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/346cae3b69e64f6bad571bdb87844ada : SUCCESS in 1h 14m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/55584d68a781467e80ce9e96148d33a2 : SUCCESS in 1h 40m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3241550aa01948f7b0990f467176ffd8 : SUCCESS in 56m 06s\n\nWarning:\n  Comments left for invalid file openstack_requirements/cmds/generate.py\n  Comments left for invalid file openstack_requirements/tests/test_update.py","accounts_in_message":[],"_revision_number":1},{"id":"932faf996da5ab58157d27e6064af27b11b4cb0c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 15:38:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fdf1e9d90e82b77cc6f7aa8982f7b397ac8a2142","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 00:43:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bca68cd04bd24cffa60b7d809e979136 : SUCCESS in 5m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5877ce64123a4773a05881e0c27e4c7a : SUCCESS in 4m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/eff664260de442d1b41bcce6b396c4f3 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60daa11d221744509e389f7a2599695b : SUCCESS in 4m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bca55e13a2c14b11b825048bccde8eef : SUCCESS in 4m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e9914a572e4a4ab58806647499635e01 : SUCCESS in 1h 09m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0a416437b03e49d2ad66a74fae9c942d : SUCCESS in 1h 30m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/99913e9ba6d3479395e8b7ef6d06d60a : SUCCESS in 56m 54s","accounts_in_message":[],"_revision_number":1},{"id":"ba380e968ab03281d86bb71fed0381f9ddffb77d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 00:43:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f97da365bcb6d17ab81c6aba7afefcd638de3ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 02:53:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8ab77fe354a4212b1d177e6ce2db3de : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/816f0be88a9d467e8ca01381c57d2225 : SUCCESS in 5m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/192a4d9b6edb4fd399f7923880e14d1a : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1513475ead2a448aaff791e1b439281b : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e59c9aa763814c099b5f2511eb1e62f6 : SUCCESS in 3m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1acc265b04904dac9627c11bc6eb14dd : SUCCESS in 1h 10m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3c6f4c54373b44a3a44f1505e7c93fc5 : TIMED_OUT in 2h 07m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/23c2ae944ed2428e9c2a89b2574560c3 : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":1},{"id":"cbf7becc00a3f095d2d89d31c79644168fe920e8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-13 02:59:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"74a9010a25d716008fb46e02dca288f46b00bcbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 05:02:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5dfd164025e4182a86864e13856fc20 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e0bef48cd59649a6a4b2cc8f91332fbd : SUCCESS in 5m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a0e97cc00d0d4d2089f6203bde0a2c4c : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26b99f80c7794881b44db29f4ae25550 : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/41873d2eb6f143c0b8ca5086e423be2f : SUCCESS in 4m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/50ceb7a3256d4fbeae45b2cdc1ba5ff8 : SUCCESS in 1h 08m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d551355327df4da1b1add1cca851a395 : SUCCESS in 1h 28m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/45acd242f702411fa12139a96176da4a : SUCCESS in 52m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7f47ef4cbe3199cfb862fcbe8240ce72fa0671c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 05:02:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eef2f698520a17945d184cf64c5975f9da683de1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 06:53:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a323c831bb6941379f683bc59ff006b7 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e36cfbcc4aed4582b995b8dd58d151f0 : SUCCESS in 5m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/983d8e14c873417b813f2294c33c09e2 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84e20dc04f89461b95c8a247b8280757 : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/57194856ca11412e94e1e5d979cc5fc6 : SUCCESS in 3m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/625c7c36cfcf406699b539d777883a4e : SUCCESS in 1h 04m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/550477a5ebfd424fb4322318b45b7fc1 : SUCCESS in 1h 19m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5322fd964bec4bc79603465119cefa5d : SUCCESS in 54m 04s","accounts_in_message":[],"_revision_number":1},{"id":"f19a94c8856fb3962e92c0579304017cd6c2bfb9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 06:53:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac3515b77f58c8ec1bf6486edc83aaa8018b5df4","revisions":{"ac3515b77f58c8ec1bf6486edc83aaa8018b5df4":{"kind":"REWORK","_number":1,"created":"2020-05-12 00:14:38.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/85/726985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/726985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/726985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/726985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/726985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/726985/1"}}},"commit":{"parents":[{"commit":"881814daf02179eb47be4f3ce248560b2b8199fd","subject":"update constraint for horizon to new release 15.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/881814daf02179eb47be4f3ce248560b2b8199fd"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-05 22:12:48.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-12 00:14:28.000000000","tz":-300},"subject":"Stop testing with pypy","message":"Stop testing with pypy\n\nChange-Id: I18928e1a7d4413423be36eeb87bbf6376ffbe2d9\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit f0fe3ce1b5771463130aad905ee2aed148884ec1)\n(cherry picked from commit 110cd69b15d49e0bcfc6726ee1ddfdc19dc8274c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac3515b77f58c8ec1bf6486edc83aaa8018b5df4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ac3515b77f58c8ec1bf6486edc83aaa8018b5df4"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
