)]}'
{"id":"openstack%2Frequirements~726990","triplet_id":"openstack%2Frequirements~stable%2Focata~I18928e1a7d4413423be36eeb87bbf6376ffbe2d9","project":"openstack/requirements","branch":"stable/ocata","hashtags":[],"change_id":"I18928e1a7d4413423be36eeb87bbf6376ffbe2d9","subject":"Stop testing with pypy","status":"MERGED","created":"2020-05-12 00:18:46.000000000","updated":"2020-05-28 23:05:11.000000000","submitted":"2020-05-28 23:05:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726990-1590707111582-48d602ed","meta_rev_id":"4d65adbd25e69249fc82667965fe791d5cd8a7da","_number":726990,"virtual_id_number":726990,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 23:05:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-05-12 00:19:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-05-12 00:19:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 00:19:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-28 23:05:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c45591904946081235ac8f5c15a8cd08438c93f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-12 00:18:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3de26d6d1dd3cbf32b397493c52e67c419c50b9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 00:19:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"809254a0bfcf77f42dd4fad6363ca8e6ce68e818","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 01:13:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0608bf2f17c7417d8de3bceeb0b84e53 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/990ee9a2961948138028c3b0f5297bd2 : SUCCESS in 4m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f8156dac3a6045128550d456c1ed384c : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c2312a85f91c4002b72869213b342e07 : SUCCESS in 4m 23s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/9a772bdbe85c46bab1729cf5015d9c9c : FAILURE in 30m 54s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/0b796107717f4538842c22275b98b72f : FAILURE in 27m 04s","accounts_in_message":[],"_revision_number":1},{"id":"06c681ae04e46f5cbc9b687ab9886c7b76d31641","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-23 20:07:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4ee390d2eb5a7b027dfd620a4643eadfa84d6320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-23 20:35:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be882971b0c4d848af9f3b306a06923 : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7f8469ff498c4b21b48aba4bb8543921 : SUCCESS in 3m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0ea38bb5816e4967903449f6d868848c : SUCCESS in 4m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f6f261e3da794cd0a11c8e44fc46bbef : SUCCESS in 3m 58s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/ea470e0f94b34821ba0c624902cf0177 : FAILURE in 22m 39s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/1041713e4d0542a585e3372ef658e641 : FAILURE in 27m 44s","accounts_in_message":[],"_revision_number":1},{"id":"fcb99a939178f848597cb4bb2dc57b206b009185","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-28 16:59:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"480d63a7f0da48c9899e1ef9be42e87fbdb4ec58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 17:15:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e4b545cea8846a0912c5f19370d987e : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6eddc68e017046b98a8b9e85ad2881ba : SUCCESS in 6m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/23152229b0334bce81dd7e6f63e04fc5 : SUCCESS in 6m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f101aaf2cbb4b0c8636545434afe9a0 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"72c2b1d1bad0a584cb5f0b1fd48d7859c6c83d24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 17:16:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5675ba23c45c1bb01dba7af3ab78f5cf81315578","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 23:05:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4d65adbd25e69249fc82667965fe791d5cd8a7da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 23:05:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b17632251874f25af9fcd5c9161feee : SUCCESS in 3m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a98b6e2f8bbd471f974ddc7c0a5b9d37 : SUCCESS in 7m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ee99307e9dd144d8985a475b49c2728f : SUCCESS in 4m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ce37636e71948478a703f24f7060a1b : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c68111538d3b1c969bccb67933084f1910276068","revisions":{"c68111538d3b1c969bccb67933084f1910276068":{"kind":"REWORK","_number":1,"created":"2020-05-12 00:18:46.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/90/726990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/726990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/726990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/726990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/726990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/726990/1"}}},"commit":{"parents":[{"commit":"831cbb687a7179d7591f29699bcd8fe09a534aeb","subject":"Retire swauth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/831cbb687a7179d7591f29699bcd8fe09a534aeb"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-05 22:12:48.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-05-12 00:18:39.000000000","tz":-300},"subject":"Stop testing with pypy","message":"Stop testing with pypy\n\nChange-Id: I18928e1a7d4413423be36eeb87bbf6376ffbe2d9\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit f0fe3ce1b5771463130aad905ee2aed148884ec1)\n(cherry picked from commit 110cd69b15d49e0bcfc6726ee1ddfdc19dc8274c)\n(cherry picked from commit af2fdfc84744d6b9899d5b5f72aa356c05d22065)\n(cherry picked from commit 55a5bdbaff68d8bac49776d9617e9e89d290e098)\n(cherry picked from commit b30560a861f93e1756364ece8bac8b6192460ad1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c68111538d3b1c969bccb67933084f1910276068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c68111538d3b1c969bccb67933084f1910276068"}]},"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
