)]}'
{"id":"openstack%2Frequirements~727002","triplet_id":"openstack%2Frequirements~master~I1000dee7f2be48f00911c9a7f5c8124d0c712e8e","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I1000dee7f2be48f00911c9a7f5c8124d0c712e8e","subject":"fix pep8 errors","status":"MERGED","created":"2020-05-12 02:18:04.000000000","updated":"2020-05-12 05:36:26.000000000","submitted":"2020-05-12 05:34:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727002-1589261699010-4b62c6ec","meta_rev_id":"6544f29950abe9ebfca58ddccd3375203d77b565","_number":727002,"virtual_id_number":727002,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 05:34:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-05-12 04:09:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-05-12 04:09:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 05:34:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6a1e37bd60ea347dad877fe8be46d2370b93ccd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 02:18:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d000fdabff8f540a99cce1492151ddc7b30f192","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 04:00:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b017cd024194fe5bc33e37f9eb6874f : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4793924da1104a2798c1c8048f55154e : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bc108c0976c41ae9a8b0bda67ff4300 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8fd040d4aa44993b6d481407af19fe2 : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4cb37d6df7fe4ab180ddfbf585bb4f4b : SUCCESS in 3m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0aaf0c7c90414bcfa29084e59d53b13c : SUCCESS in 48m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/426d329f6cf74db7894e3af6ff66ee4e : SUCCESS in 1h 09m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec5dbc5964ab41df94bf33671aac7894 : SUCCESS in 1h 37m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/cda9fb85b84d452b82106f65368c03aa : SUCCESS in 1h 14m 41s","accounts_in_message":[],"_revision_number":1},{"id":"c8372b44e0e7145697c255f17334925edacec423","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 04:09:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6bfb3f79b687cffe27b38471a154ec23a3ae2124","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 04:09:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ada8797a37348ee5ae191aed9cd7102a763929e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 05:34:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/395b569e7da94ed1afc7bccbec5fc91c : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/389b93b9398a4054b609193a75e58313 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77a42bf51bf74320ab3687e751a31af5 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5627caa807440a8851345fc4666ec70 : SUCCESS in 5m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df9ef2403550473489831f11268d47b1 : SUCCESS in 4m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6172cf5cadd749398351041ff93120de : SUCCESS in 50m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9676e982e00b4cbd84d3aa53e6380aeb : SUCCESS in 1h 06m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f4469945a914bc0b9402fa1fbf215ad : SUCCESS in 1h 22m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/1a6801a8edda449ebc0be07c4e5fc0ab : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":1},{"id":"aae2d536c508a492f03886331d7def4e6fdfdd6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 05:34:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6544f29950abe9ebfca58ddccd3375203d77b565","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 05:36:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255b17b4e8e2424ca4d7b612e3e6c47a : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d92bd96ee1f20e829610511d24131f1888ddf37e","revisions":{"d92bd96ee1f20e829610511d24131f1888ddf37e":{"kind":"REWORK","_number":1,"created":"2020-05-12 02:18:04.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/02/727002/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/727002/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/727002/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/727002/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/727002/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/727002/1"}}},"commit":{"parents":[{"commit":"aeeaafe47d912e5ab2b6266b2e18d28321d31c36","subject":"Merge \"Block influxdb\u003d\u003d5.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeeaafe47d912e5ab2b6266b2e18d28321d31c36"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-12 02:17:50.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-12 02:17:50.000000000","tz":-300},"subject":"fix pep8 errors","message":"fix pep8 errors\n\nChange-Id: I1000dee7f2be48f00911c9a7f5c8124d0c712e8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d92bd96ee1f20e829610511d24131f1888ddf37e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d92bd96ee1f20e829610511d24131f1888ddf37e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
