)]}'
{"id":"openstack%2Frequirements~727052","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I1000dee7f2be48f00911c9a7f5c8124d0c712e8e","project":"openstack/requirements","branch":"stable/ussuri","hashtags":[],"change_id":"I1000dee7f2be48f00911c9a7f5c8124d0c712e8e","subject":"fix pep8 errors","status":"MERGED","created":"2020-05-12 05:49:38.000000000","updated":"2020-05-12 10:45:29.000000000","submitted":"2020-05-12 10:45:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727052-1589280329697-61dee02a","meta_rev_id":"7894b8beb2d0f337acd10bae49e7754696ec4a57","_number":727052,"virtual_id_number":727052,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 10:45:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-12 08:49:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-05-12 08:49:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 08:49:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-12 10:45:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5261d28d85f2e2d4138b14302e09dc153c39281","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 05:49:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20ed746e1d112de2e9efb907df0408ab50a2c8b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 07:15:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07da794baa09435a8d3a89aaf071277d : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6eab5047cda842399b48e8ddf1b64d2f : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ccdc3dd16474474b9d04fc2b5f6128d6 : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1bc3629448334a459c97a4fd8bf17b3c : SUCCESS in 4m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e14f6a7f68446729d7e5cd4309b35ff : SUCCESS in 53m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/949bb519c40c457d89c1f042ac69e18c : SUCCESS in 1h 18m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55158a7099634913b2aace1e7cd15c0a : SUCCESS in 1h 25m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/77ad873f88534a3b8be7c5166f2f14bb : SUCCESS in 1h 15m 22s","accounts_in_message":[],"_revision_number":1},{"id":"e41f8079663e0a759138df32637575b152254d31","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-12 08:49:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"81e42602366c8ed363c786c33510c943ce60faf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 08:49:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9eed121ba858470bdcb1b4e4edd32b38d9b6a073","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 10:45:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7894b8beb2d0f337acd10bae49e7754696ec4a57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 10:45:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dffeb8e0004e4483b7903625839124db : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/137e6f8b67f04078a7f43c7bfd94992b : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/299337ef56d34a469c1dfdbb6ee62b07 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/493347f1f95043e4875980babcad34e4 : SUCCESS in 5m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/441addd2fb7a42ef8477e686b7333399 : SUCCESS in 59m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6f2dec66ccc4f9e91ecada7ca2eee5d : SUCCESS in 1h 20m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/260e9c5e6a0949aaa26bff07bdeac031 : SUCCESS in 1h 30m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/e129acf5e67645abb8da1411d98f3ca6 : SUCCESS in 1h 05m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6bd7c338b9bc425ba87418887886f822f881872b","revisions":{"6bd7c338b9bc425ba87418887886f822f881872b":{"kind":"REWORK","_number":1,"created":"2020-05-12 05:49:38.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/52/727052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/727052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/727052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/727052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/727052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/727052/1"}}},"commit":{"parents":[{"commit":"7d3a23d79bea8a66890751f1c123db6e04514e72","subject":"Switch to new grenade job name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d3a23d79bea8a66890751f1c123db6e04514e72"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-12 02:17:50.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-12 05:49:29.000000000","tz":-300},"subject":"fix pep8 errors","message":"fix pep8 errors\n\nChange-Id: I1000dee7f2be48f00911c9a7f5c8124d0c712e8e\n(cherry picked from commit d92bd96ee1f20e829610511d24131f1888ddf37e)\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6bd7c338b9bc425ba87418887886f822f881872b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6bd7c338b9bc425ba87418887886f822f881872b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
