)]}'
{"id":"openstack%2Frequirements~727053","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I1000dee7f2be48f00911c9a7f5c8124d0c712e8e","project":"openstack/requirements","branch":"stable/train","hashtags":[],"change_id":"I1000dee7f2be48f00911c9a7f5c8124d0c712e8e","subject":"fix pep8 errors","status":"MERGED","created":"2020-05-12 05:49:48.000000000","updated":"2020-05-13 02:07:48.000000000","submitted":"2020-05-13 02:07:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727053-1589335668644-c4901b80","meta_rev_id":"eb096f8ef66d89d71902a051267a2509d1d06130","_number":727053,"virtual_id_number":727053,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-13 02:07:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-12 08:49:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-05-12 08:49:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 08:49:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-13 02:07:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e66180487f1bce93c11189d8408bb5c3bb6a940c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 05:49:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcaef0a289fdffb2d438c9e1db5f83a65916e750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 07:32:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/08a3bd7980534fabb887fd220ddf1325 : FAILURE in 3m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b625d179a8b4305988e96236aac6f53 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e04c5f424714f6887a0296c708eb176 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51aa0b273ea64df6aa0d75235770d78a : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ef8cdb0466b409091343bdf03f3d96d : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9aee5a7af4e64e998fd3265e551489c0 : SUCCESS in 3m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d40db81eefd463e9889516b23a80c5a : SUCCESS in 1h 17m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/af7176d6edfe445daac50bf51cc1ff17 : SUCCESS in 1h 37m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c9777c66bb6f4ae0bcb5e9a776a0ac4e : SUCCESS in 57m 26s","accounts_in_message":[],"_revision_number":1},{"id":"80f63a8ee9884e4f30f01b7207bfa625215e4cab","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-12 08:49:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"38751f295735dbb29c13dbc9ee36711c5b84ec45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 08:49:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5ab5ff3ba7cf11245aa7c0457a55e19992b86256","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 10:45:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d52813288054eb2920dc15d0a8576f6 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e5f0b6925b804ef18f70046b82767fcc : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97f155d52b604e65a248a94d0d43024c : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c6bf1cccd2b4dcf8223b7074f364522 : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a822449cc1504016a1668e754331a88b : SUCCESS in 4m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed01789d1e644f22b7f121269429e707 : FAILURE in 11m 44s\n- tempest-full https://zuul.opendev.org/t/openstack/build/09066c63b19444f39e2ba3c026cd0e33 : SUCCESS in 1h 24m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f55c939c3205477f87dd70da3d6dc803 : SUCCESS in 52m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ab183380e84672e8e22b36005135f4e8f85abdce","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-12 15:01:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3fd0be9e00ba1e76b45b7412dbab1be0b6ca614a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 00:04:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/e23aaf552e8f4151887f31de18216762 : FAILURE in 5m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9fdbe2dbfff49bc9c266aaa896dac74 : SUCCESS in 6m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7898bf31ce247afaabda45aa46f3977 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f0e9a81bdc04863b0f03e02f3dbd1fb : SUCCESS in 11m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b14f00d6a5224399bf5cce82ba8dbc05 : SUCCESS in 10m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d4f2c8c57ea4643864ac3fc57f2fd82 : SUCCESS in 5m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/29642194343e438ea9b8b93800253e2c : SUCCESS in 2h 15m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b292162807d04dfaabec881c9ebf8222 : SUCCESS in 1h 37m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e62543a1a4be44fbbf0bf6dda22278b9 : SUCCESS in 1h 05m 37s","accounts_in_message":[],"_revision_number":1},{"id":"9c3c85210f7b6a999385150419a98248a2cd365f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 00:04:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bcbf45446bef393c176e2987141abcd7843f5d9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 02:07:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"eb096f8ef66d89d71902a051267a2509d1d06130","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 02:07:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b15ecbdcc8748ccb1b7773c51af16cc : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b4be3a49bf9347a388be2c8d7d02037a : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a8900eb44e849b69893989b46389c6d : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/147070bc55654717964294a04c460935 : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ec5230287b11468fb1943425e3078bd0 : SUCCESS in 5m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a7b1687ac334444b8334a21954b8677 : SUCCESS in 1h 07m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f31863d2661d44039fd987ab43dbf6b2 : SUCCESS in 1h 56m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ee2b7720e36d4a379429b940958a4a2b : SUCCESS in 1h 05m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1fb8b040c35e97477d06ee3f159cb1b773965c6d","revisions":{"1fb8b040c35e97477d06ee3f159cb1b773965c6d":{"kind":"REWORK","_number":1,"created":"2020-05-12 05:49:48.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/53/727053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/727053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/727053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/727053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/727053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/727053/1"}}},"commit":{"parents":[{"commit":"311b2b41f693e656ee38ced0b9868fdf7c148b22","subject":"update constraint for castellan to new release 1.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/311b2b41f693e656ee38ced0b9868fdf7c148b22"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-12 02:17:50.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-12 05:49:43.000000000","tz":-300},"subject":"fix pep8 errors","message":"fix pep8 errors\n\nChange-Id: I1000dee7f2be48f00911c9a7f5c8124d0c712e8e\n(cherry picked from commit d92bd96ee1f20e829610511d24131f1888ddf37e)\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1fb8b040c35e97477d06ee3f159cb1b773965c6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1fb8b040c35e97477d06ee3f159cb1b773965c6d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
