)]}'
{"id":"openstack%2Frequirements~727058","triplet_id":"openstack%2Frequirements~stable%2Fpike~I1000dee7f2be48f00911c9a7f5c8124d0c712e8e","project":"openstack/requirements","branch":"stable/pike","hashtags":[],"change_id":"I1000dee7f2be48f00911c9a7f5c8124d0c712e8e","subject":"fix pep8 errors","status":"ABANDONED","created":"2020-05-12 05:55:23.000000000","updated":"2020-05-12 15:52:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a41da44f5c8c20c418012257d5cfa555bb66f429","_number":727058,"virtual_id_number":727058,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-12 07:27:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 07:27:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac88c0b61762f63d0f4527a488b5bb82a20c357b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 05:55:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43591216461a54579db71ce1f6a119aa0e8daad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 07:27:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/7c9415a9e96342cc9bf1bbff801faba3 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96d821e903a74f52943839eb45c72e25 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/abb4d66b5c4e43e28784ce73eb768616 : SUCCESS in 4m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/747dcdcaba25438ebd919b1668099d40 : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3aa8abdef97848a9aebf46b282403ed3 : SUCCESS in 5m 01s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/4e8c5e2b9e5a46889a9016392778e74d : FAILURE in 38m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dbbddeb5c7324ddf99bb4eed799947f8 : FAILURE in 36m 20s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/4f68161e0ece400ea1f8d83e40262bc3 : SUCCESS in 1h 27m 22s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/59d451d3381d4f3d92fc1bee133ee09e : FAILURE in 33m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a41da44f5c8c20c418012257d5cfa555bb66f429","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-12 15:52:04.000000000","message":"Abandoned\n\nuse https://review.opendev.org/726982","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d347706f2cb4145bd40e11a39718de8dcaf5642","revisions":{"4d347706f2cb4145bd40e11a39718de8dcaf5642":{"kind":"REWORK","_number":1,"created":"2020-05-12 05:55:23.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/58/727058/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/58/727058/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/727058/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/727058/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/727058/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/58/727058/1"}}},"commit":{"parents":[{"commit":"36c442553e29181c69fcd730c3cb145e3822adb0","subject":"demarcate python-memcached requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36c442553e29181c69fcd730c3cb145e3822adb0"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-12 02:17:50.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-12 05:55:15.000000000","tz":-300},"subject":"fix pep8 errors","message":"fix pep8 errors\n\nChange-Id: I1000dee7f2be48f00911c9a7f5c8124d0c712e8e\n(cherry picked from commit d92bd96ee1f20e829610511d24131f1888ddf37e)\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d347706f2cb4145bd40e11a39718de8dcaf5642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d347706f2cb4145bd40e11a39718de8dcaf5642"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
