)]}'
{"id":"openstack%2Frequirements~727694","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I7f9e8690c2e24464b26aa629a30fe306bc190c59","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I7f9e8690c2e24464b26aa629a30fe306bc190c59","subject":"update constraint for openstack-release-test to new release 3.0.1","status":"MERGED","created":"2020-05-13 09:50:41.000000000","updated":"2020-05-13 17:14:38.000000000","submitted":"2020-05-13 17:14:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727694-1589390078416-03b7720d","meta_rev_id":"7011a310f06794e44fcc4fa22ad2ff32a9670f83","_number":727694,"virtual_id_number":727694,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-13 17:14:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-05-13 15:06:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-05-13 15:06:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 15:06:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-13 17:14:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"06d7db09edff60b86ccafed6a7e9afd1a1a8c6c1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-13 09:50:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c590f76ae3b080205d422cb2a534a0ab5df5533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 11:39:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce6f794de00455dac6bd9e83be7324a : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de0aeb27b66d4ee28b11087aec53374b : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78a2a9bf962645d295c45e5b4576dc80 : SUCCESS in 4m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a747085e4891450d85ae1ec39b13ebfb : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a3f00527c795445694341bd3725b8563 : SUCCESS in 3m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3ccdba7b4d8e4af8bb0e7130e130a0e3 : SUCCESS in 9m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4d6bd7b861194111b59aaf7ae6eb33c5 : SUCCESS in 8m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/11940c37d1d34df19891cb4276941fa0 : SUCCESS in 5m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cb2cead10ee941d58535a06367da100c : SUCCESS in 12m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7ee774767f7f46a9adf59f7054579d33 : SUCCESS in 7m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a19fcfabb6a349d9a6975229c8524e25 : SUCCESS in 13m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/839431bb24ff4f85b4e6051e25a1074f : SUCCESS in 29m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2a3a3fb6531547d998bb79c75e442fb0 : SUCCESS in 14m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/0d6635afb02c474f9366810b5541d4cc : SUCCESS in 15m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cafb51c985144c02ac865fb388607e9e : SUCCESS in 18m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/73ba94178a5547639dd02be257dd9d60 : SUCCESS in 5m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/24b318fee8504648bbd9ac3d868f8d45 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9062451c6c9744dbb15d04d2c6e5de81 : SUCCESS in 7m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2fe7ff918bf441c0b0ffc3aad54fdef6 : SUCCESS in 9m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f9dfb8ee30b14530b985a64803bbd579 : SUCCESS in 11m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3687284c882043a7b0c2a91360e69389 : SUCCESS in 1h 02m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ade7bcb9a09429f8342339ee511ecd7 : SUCCESS in 1h 17m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb8640f281b44d2583b8ab8f43d4ece8 : SUCCESS in 1h 19m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/68d2a8b529e9400483a7a5d9403bd652 : SUCCESS in 1h 14m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a23d4d3b760a4658b6b29911ea45ee7c7d115859","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-13 15:06:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7fa66cbcdc51a4bc09fe2d21697437a59a28966c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 15:07:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4a4c808f1b39212dc7172dbc9c855280acd147a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 17:14:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"30e84a6c11c9011b71f23027f2c6720b2ba1a8f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 17:14:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/688eeac21b1e4d9cbe14d2b687267b0c : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55f276ce43704fca86418d3433fc2c61 : SUCCESS in 6m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7030784f8a24099be133683f6381517 : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50022aa023134c0aa856aac9d81d704f : SUCCESS in 5m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/34169953c375423eaf31fc78d7d513c1 : SUCCESS in 5m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a5d6bac87726417baaeac689abb0081e : SUCCESS in 7m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6d8050b78cb243f6a18d5c431b52bb75 : SUCCESS in 9m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d2c76912a7a8480ab383f3c40ce1a4f3 : SUCCESS in 5m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a8ee270fe0ea456781cd1f1c1cf2f1c8 : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5c52b653ca3e4d5987ce3fa62be1af43 : SUCCESS in 8m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f7189becfe5b4a31951ec7f3e6be18fa : SUCCESS in 17m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0694dfdc869e4f009ce0f97a369a9314 : SUCCESS in 39m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f790421ec0ab44deb83591bdcdaa70e4 : SUCCESS in 12m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/cea2ae4e585c49df873e34d306301f39 : SUCCESS in 13m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d3cccf0d882d4015938364cae8d410e5 : SUCCESS in 19m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/65fdb667574d454a9ec8d034b241a3d1 : SUCCESS in 6m 47s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8b173f6c36764524bdd79e7b10228ce5 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cd2e4ae9aabe4844b793aae0ebfbef67 : SUCCESS in 6m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b8cd70af49ad4e0aa69f37edb24fd36b : SUCCESS in 9m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a77e90961b9f4b60951615151e58ac83 : SUCCESS in 12m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62c60d27368f431c82996c43b77ca788 : SUCCESS in 52m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4944518366f4dd588e56f3ec18ee1da : SUCCESS in 1h 09m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ffdf98d2be804f7b9bae50ef6c482d21 : SUCCESS in 1h 12m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/6c52b7f708264119a7a0b2d68ed88110 : SUCCESS in 1h 17m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"991612e6ce38a2cd8f211f787bc41888cb3db940","revisions":{"991612e6ce38a2cd8f211f787bc41888cb3db940":{"kind":"REWORK","_number":1,"created":"2020-05-13 09:50:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/727694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/727694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/727694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/727694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/727694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/727694/1"}}},"commit":{"parents":[{"commit":"6bd7c338b9bc425ba87418887886f822f881872b","subject":"fix pep8 errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6bd7c338b9bc425ba87418887886f822f881872b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-13 09:50:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-13 09:50:40.000000000","tz":0},"subject":"update constraint for openstack-release-test to new release 3.0.1","message":"update constraint for openstack-release-test to new release 3.0.1\n\nChange-Id: I7f9e8690c2e24464b26aa629a30fe306bc190c59\nmeta:version: 3.0.1\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Change-Id: I8b9be10e7a090bdf87b205cacae76f8d62621800\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/991612e6ce38a2cd8f211f787bc41888cb3db940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/991612e6ce38a2cd8f211f787bc41888cb3db940"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
