)]}'
{"id":"openstack%2Frequirements~728018","triplet_id":"openstack%2Frequirements~master~Ibf26d927cd45d0f39a0fcb7c4700786dbf002459","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Ibf26d927cd45d0f39a0fcb7c4700786dbf002459","subject":"Add yamllint to blacklist","status":"MERGED","created":"2020-05-14 10:31:09.000000000","updated":"2020-08-19 12:53:15.000000000","submitted":"2020-05-14 17:04:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728018-1589475847385-eaedcf2a","meta_rev_id":"1ff89ca2241018f158ba196729adeb20325508ff","_number":728018,"virtual_id_number":728018,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-14 17:04:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-14 11:40:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-14 15:36:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-05-14 11:00:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-14 15:36:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-14 11:00:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-14 11:40:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-14 15:36:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-14 17:04:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cbad30ea06a281f46f7ef9776b42eb98d8824d27","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-05-14 10:31:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2b4968c10340453e25950aab9cc2bacba974e1e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-05-14 10:32:36.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5a5327752763d1febb4b6b54c7833fd2c0d1a982","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-14 11:00:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8b8f0aa7c49e37b682ac42bffef0ec269b5d5a7b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-14 11:40:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8c91219457b43fc6652536176abdc5aedee7fa0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 11:51:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43ba1b39291d419d83907da953c3ff8b : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1ba2f5a60c84df4b938268d157132fa : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30d3091acd194652bf71562bdcd8ef00 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b03a2134b0e647a3b00c2a932b20d0a0 : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca4cde5cdded4330a3449af499f083fe : SUCCESS in 3m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00eea4e24dd14b73ad0be02f3d520402 : SUCCESS in 50m 12s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/92cab9d87fae4351a8b979f292b79159 : SUCCESS in 1h 07m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eef61b1096cb475cab742f7ea9e0cd2d : SUCCESS in 1h 17m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/df8af876518f4c2194221ddf2f71c3a6 : SUCCESS in 1h 12m 04s","accounts_in_message":[],"_revision_number":2},{"id":"b7209b240462f366997d54dba0736650355d60d5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-14 15:36:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cf94aa35a1260fea01f60da2c917eceb09309762","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 15:36:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a4ab266deadc7f9314c4ea35634c2d9a162b2bb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 17:04:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba9f1cf56a384625814c55282fc6339f : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/592ba3c8a4a846479188e6b2172a0f68 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1c842fad1454e4391124236d657e70c : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5584c759945f47a281f5968f1242ec78 : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2cd1b11fb8914d69a9a633532546e9c5 : SUCCESS in 4m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10e20533dad04e04b9f2f00d7035a8b3 : SUCCESS in 1h 02m 33s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/604b6d8e317d423590117e7bee5086ef : SUCCESS in 1h 21m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a99f7b6fbbf46aa8fb126aaf031c862 : SUCCESS in 1h 15m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/8e947eef46934e6d99021c037955a070 : SUCCESS in 1h 17m 49s","accounts_in_message":[],"_revision_number":2},{"id":"8ce19f9e38ccd6b230bcf2406ecced73d1e269a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 17:04:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2d0cf3753cd2ca4c7b0aeb5a4659b5bd48141976","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 17:06:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62338867ef914d1886e936a196552413 : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":2},{"id":"1ff89ca2241018f158ba196729adeb20325508ff","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-19 12:53:15.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 05f7a31434845ce8b33a7e44c0e2c6aea97dc150","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6ca51f2b1c52a6691925029586b6514ede8f13a1","revisions":{"b5eac333f52df7835f800ebbf8c074669ca1beb9":{"kind":"REWORK","_number":1,"created":"2020-05-14 10:31:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/728018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/728018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/728018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/728018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/728018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/728018/1"}}},"commit":{"parents":[{"commit":"9500cbb4ecd0a1bf0848f4209d28ce1175cf1e0d","subject":"Merge \"update constraint for openstackdocstheme to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9500cbb4ecd0a1bf0848f4209d28ce1175cf1e0d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-05-14 10:26:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-05-14 10:31:06.000000000","tz":120},"subject":"Add yamllint to blacklist","message":"Add yamllint to blacklist\n\nSome projects, such as kolla-ansible, use yamllint for linting YAML code.\nAt present it is not possible to add this to your requirements as it does\nnot appear in the global requirements file.\n\nChange-Id: Ibf26d927cd45d0f39a0fcb7c4700786dbf002459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5eac333f52df7835f800ebbf8c074669ca1beb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5eac333f52df7835f800ebbf8c074669ca1beb9"}]},"branch":"refs/heads/master"},"6ca51f2b1c52a6691925029586b6514ede8f13a1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-14 10:32:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/18/728018/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/18/728018/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/728018/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/728018/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/18/728018/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/18/728018/2"}}},"commit":{"parents":[{"commit":"9500cbb4ecd0a1bf0848f4209d28ce1175cf1e0d","subject":"Merge \"update constraint for openstackdocstheme to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9500cbb4ecd0a1bf0848f4209d28ce1175cf1e0d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-05-14 10:26:59.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-05-14 10:31:42.000000000","tz":120},"subject":"Add yamllint to blacklist","message":"Add yamllint to blacklist\n\nSome projects, such as kolla-ansible, use yamllint for linting YAML\ncode.\nAt present it is not possible to add this to your requirements as it\ndoes not appear in the global requirements file.\n\nChange-Id: Ibf26d927cd45d0f39a0fcb7c4700786dbf002459\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ca51f2b1c52a6691925029586b6514ede8f13a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ca51f2b1c52a6691925029586b6514ede8f13a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
