)]}'
{"id":"openstack%2Frequirements~728367","triplet_id":"openstack%2Frequirements~master~Ibab788d0b31e2e1da145f20710233419b982b4c2","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ibab788d0b31e2e1da145f20710233419b982b4c2","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-05-15 06:16:53.000000000","updated":"2020-05-18 09:08:24.000000000","submitted":"2020-05-16 22:40:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728367-1589668831904-be6e565f","meta_rev_id":"dcb3c3c1b399146bbc73321ac45d8bf96480a382","_number":728367,"virtual_id_number":728367,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-16 22:40:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-16 21:02:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-16 21:20:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-16 21:20:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-16 21:02:26.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-16 21:20:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-16 22:40:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec096abb7c943084f284febe637ac9c32bcf7c3e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-15 06:16:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4197b59a8bb05dee7d57f3c61c7ff8bdf1627886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 08:15:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/118718d58afa4cd69d47f39def7dcfe1 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6278f822636b45caa5cf47c65c0b20e7 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0f6966323c846feb69c2bb42f846b12 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0efc3a447aba4d4f94214eafe2ab1ba2 : SUCCESS in 7m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ee587572a5a24a428aeb32a4653afe0f : SUCCESS in 7m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/34162acfef9c4e33be671bbc57e61ca8 : SUCCESS in 6m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3ca012269975462085120cb5d232fec8 : FAILURE in 9m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/deb48824cb554a6ca2979e20f58939a6 : FAILURE in 9m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/372f22bb4ffa4269ada72c3c02ab0871 : SUCCESS in 6m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f8217d8446fd41bf8d06d3807ad1d571 : SUCCESS in 12m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3c28bfe873f94252aa79d73ab02f11e7 : SUCCESS in 11m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d7136b18ef9546fbb12abc0406e0756f : SUCCESS in 48m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/540b5dc7d6c448da8c2ff0957759d3fe : SUCCESS in 27m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0df19061e27b4dc49dacf49c5a4bc386 : FAILURE in 16m 20s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b87aedf200cc419caad38d351d966e85 : FAILURE in 15m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3ffd3620875640d19315e25612bb7cdc : SUCCESS in 19m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/62ae2e9b51964b389e91da828ab0c418 : SUCCESS in 8m 30s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9f4dac629e4b44b2bf571bd1d6f0fe6d : SUCCESS in 7m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f6912d5edb8a46cf8e420d69d49e6605 : SUCCESS in 8m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/218d992f7f5944e1b275e933e84e8d0c : SUCCESS in 12m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4f79b823bf684d87942d25a8db78decf : SUCCESS in 14m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38f3b310a04e4afbaca99d8bd2446ce8 : SUCCESS in 56m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/164256e93966462285fbe8ccf416b8aa : FAILURE in 1h 12m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/035cbf682ea842ffbe690d660feb3c74 : SUCCESS in 1h 20m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/0054abc6e1d94926837b10dc14fb8f55 : SUCCESS in 1h 09m 24s","accounts_in_message":[],"_revision_number":1},{"id":"83d28c7a269e7df6fbd435b848a5bd833ff1c5c8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-16 06:13:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0adaa4a90c6cd22aad7d20dc0ef4e009dc7e3e41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 07:55:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c11463468d904b3a893a8683f2b3afb8 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fea3c1fabfe4422a5284b726eb426de : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/244e7330e3584c7083094e880eb2f2f9 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0fb7c701e574b7ca5ed68371d8c4b48 : SUCCESS in 7m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/942f3c591c304c76a11b31438104f539 : SUCCESS in 7m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf916c5984354547ae03200dcc402b17 : SUCCESS in 6m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a44b6147b55d482697fdd7f848f3abca : FAILURE in 10m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1204481187354327baabd875b4f04443 : FAILURE in 5m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e84d96abe9034da59dbb07f3032c193e : FAILURE in 10m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8b5f2a818ede4b6a94f4c709d652c51c : SUCCESS in 7m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d570e12da0af4f6ea2f0fefe06ddc7cd : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/58cfa4f92dab4c84b497fb05bb19fc79 : SUCCESS in 10m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/244de04f832746b49a3fbd0f19f777cd : SUCCESS in 15m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b7bfb72f4d5d46afb3b90e04ec163024 : SUCCESS in 24m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ae4a8ee6b1b044998e1f000406c088c0 : FAILURE in 15m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c20332e6a3a049d69fde8ad3aba17eee : SUCCESS in 17m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4c26afbd4c2f44d4a93f3a1a0b0ba0fb : SUCCESS in 7m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a6d2cfa4d9d4c9a9d11999bc09f7105 : SUCCESS in 7m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9c58be385f334e57a700e636603db037 : SUCCESS in 11m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0b99e9232544af586d2773e86a2dde1 : SUCCESS in 46m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8069a379d5204ea0b073d8710e8edbfd : FAILURE in 1h 06m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75b27de618194bd9a3ca06109d9e171d : SUCCESS in 1h 13m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/bb4ce718fc754446b95592fda988c033 : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":2},{"id":"3e5264cf80e899b9ed33e5a48396518699d2c25d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-16 15:53:33.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"9e5cd6199276cb23f3289681f7b578d0102cf5e2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-16 15:57:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0a32bc0778614d76429b626c82233eea9c33a724","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-16 16:37:56.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4bd93602364b808159d40baf14f18a5c4003ec0f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-16 16:46:48.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e231df7a3df5767a64d6100318590cbc18926d5b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-16 16:58:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"69fb5dccb73958925628484d4ba1dad5470d6f3c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-16 17:07:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"329e3c9473c6016457eda9db0ec67c33ebb577f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 17:12:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26b8d7ce767e4dceb94abec130f081e6 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add6101fab5d43cbaf3ef493b3e9c195 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0136644319aa4b5995229bc55a93a242 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dedcdff5f6974e1faba6fd97c0b46c44 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/104ed055c36b418c8e7821645e04ab51 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d12016df38c8484097cfd1400a6868aa : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fdb2d793e87a4ed48443396a5be5fd58 : SUCCESS in 8m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/579def537f67495281f17c4ef5f1eddc : SUCCESS in 9m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/224819c78a1045cc99c962d9d575d12b : SUCCESS in 9m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e1c4b6e0e69d4a669f1820f45a9b4e4e : SUCCESS in 4m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3efddfbe067a49e499bb4c0dcafd474e : SUCCESS in 8m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b68b44756f5047bca20cda1b870bb236 : SUCCESS in 7m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/054f001a5a0e4edba1e8f8b8ff6fe60c : SUCCESS in 13m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2015d07368e3418184806b850a48e74c : SUCCESS in 22m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5b036679fe4546698d6fd4fdb6ce6658 : FAILURE in 13m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7aebd8f59e1c4dad94e4fa4eb875cf8a : SUCCESS in 15m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6f67e447f21a41358a0d3eecd0d6b69f : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ce4c450cb9a4130b4b2bbc685168d33 : SUCCESS in 6m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f19e36a283994af2b554f31eb565bf68 : SUCCESS in 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b769ea8276884f2d985a2a9fa7f0a827 : SUCCESS in 51m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d28d9d39d2734666a973b6f21814a675 : SUCCESS in 1h 13m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1788d64a7331466182f7771ba323673d : SUCCESS in 1h 14m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/352c285a4bfc4a299edff7bdd96ac6a9 : SUCCESS in 1h 03m 15s","accounts_in_message":[],"_revision_number":3},{"id":"dd2c94c40e6a2bb67f6c809c7b62799276ae6d86","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-16 18:04:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"172a003cb4fc730e4fe21ea53ea3968818365fda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 19:28:46.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6fd8933eae54686a9f57654064ced76 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a8485213300473db6571c367543cfd8 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f20ad32ea223440fa53f14343003fcd2 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f592f7d5623440ffa6abaf47feb19da1 : SUCCESS in 3m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/238a5419eeb246e5b9047c4bec7a03db : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fcb61e34b2a04ec0a823086e3405b8ac : SUCCESS in 2m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c184d1a65d7e4942a307ab56f69a8907 : SUCCESS in 7m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/13e2750b9114496da4a90330ab66bac6 : SUCCESS in 8m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b606f54543574a58afb5a7d63d4c93c5 : SUCCESS in 7m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3cbb157eaf8b4661a390e01c88ecbf36 : SUCCESS in 4m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b39e772bb42246d49f319bc68017d2f3 : SUCCESS in 10m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2f1da5de54aa47e7a0fd477384898bbe : SUCCESS in 7m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/98e2b0c8ec234ce7946942328abf81b6 : SUCCESS in 12m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3f028d8154b94db0862434ddd0193ebd : SUCCESS in 29m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3e4fcc1799834ab1ad9ad651fb9b7404 : FAILURE in 12m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82522f8db15b4c448fc76a1457ed88cb : SUCCESS in 16m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/03ac09ee04ed409fa2dcb1ff378005ef : SUCCESS in 4m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be353d0bbb924a13b4838eb4f15e1cfb : SUCCESS in 6m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/57cbdef6b49a41fe9e9e597867dde6be : SUCCESS in 11m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/696266f0b5ef471e9bfb525436b0404b : SUCCESS in 48m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5890af7bf1414f73b9190f3a12282938 : SUCCESS in 1h 04m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66a795f2429040baa4e7b039783123b4 : SUCCESS in 1h 24m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/750264ab9526428495b6cfc8b886f2b9 : SUCCESS in 1h 03m 39s","accounts_in_message":[],"_revision_number":3},{"id":"83c49ec5900eefc89a5a163f871bafe642109a9f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-16 19:40:01.000000000","message":"Patch Set 3:\n\noh, mock...","accounts_in_message":[],"_revision_number":3},{"id":"5be4982b7c6b4cbbfd9bf77b8fef4584349b1bf0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-16 19:40:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"161d9c008a5fd3a19d4141a676890f83bca45b29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 20:56:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dd30c93ca614be88d95d9f3ec0b048f : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04639f3f947a47768ccc8c8ee5d2f669 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d04e38cc06c480b8ff9164b573e8e93 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7698138ea96e48dba22a04028c102fb0 : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5d62dbd31a540dbbd9703a516296125 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/220dda9aa8fc4451ace2465c744d983f : SUCCESS in 3m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7d65b1b596d14bb3bc7a5845a67df58f : SUCCESS in 8m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fbd67126209f4028957b4b6dd671695d : SUCCESS in 9m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d0ca1a57b6304f2ab3ad34df0b21fbbd : SUCCESS in 9m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c4cae7ad2fef44d09aa4373cf9f5f37c : SUCCESS in 4m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4563d9d00ea9460e9fa5d5feec6ca4f6 : SUCCESS in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5c05c74191cf489aa9f01ab51e157da4 : SUCCESS in 7m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f37a58f5de8d4ac0bb4e761581fe3e67 : SUCCESS in 24m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ea9beb8e18b947c0ba571d1496f8daf1 : SUCCESS in 27m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9e2280d0dbdc46c3bc52b6bb295627a8 : SUCCESS in 12m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/696944981b554069b02be9004a6cc650 : SUCCESS in 15m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/153e3b973e9a450d8071073400ad5d0f : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ac3549f764064af5be6dc9cb584bda25 : SUCCESS in 6m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9053bb638c4a4fbf8be399af9c6cd8f6 : SUCCESS in 10m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a687325355a64491ab3c81036bc27c62 : SUCCESS in 47m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fff18fc2e7d546d6ad457fce6ddb0cec : SUCCESS in 1h 00m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be63c0c869b94a769983eeaa0f402eed : SUCCESS in 1h 14m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/b669fb39b94a4bcab025e9010fd9e805 : SUCCESS in 1h 04m 03s","accounts_in_message":[],"_revision_number":4},{"id":"2e300ccac4521fb1dd5f5cefc1384769e4a6ae34","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-16 21:02:26.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7890312fa229b1bcd034de8f78332889f77ec768","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-16 21:20:34.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f4425982c84498d883e5ddcfe1910beca9007bf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 21:20:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"62b3f24d3fc774b25ba44ca7687bbbb8fb5eaa9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 22:40:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"baf7e6916818de28a4aae9766a4fcc44b3dcafee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 22:40:31.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac0c537766dc40e9ac4772b29ab535c2 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e708ac688ac43278c132884048070c6 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95a4f00e865547eaa671e1bfb1fc3f98 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a750d9d5502c4f698e633889377a1ced : SUCCESS in 4m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ba212feb577447c2af9d569d001699d3 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/121f20f0bc384245ac43c212bc870b6e : SUCCESS in 3m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a276f8f2c99448bf91d360887fa6edc9 : SUCCESS in 8m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1719e18c4df9459b9e628578138f584e : SUCCESS in 9m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/26ae7281043e491cb4586a6ad7ee98b8 : SUCCESS in 9m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c75081ca918643e39535482f70b90020 : SUCCESS in 5m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/25cb9010b70644a9bd0d02527fecaecb : SUCCESS in 8m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1df7e7e8de914364aee401fc2350036c : SUCCESS in 9m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8a74d34ee7404f58b3b3abbda6752c9a : SUCCESS in 22m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ec60faaa86cf49bd80c50337a3ce726f : SUCCESS in 28m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d44073bfca1444b480dc19ad6e543467 : SUCCESS in 16m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d684aca1753d4736bbb9cb10be2ddbd3 : SUCCESS in 17m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/59a7b9e467fe40f6a8f59bba75f374c9 : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a9a5f66b20744d0f8d925ef08164eeed : SUCCESS in 7m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/287a16217f214b6ab020fcb94e6b8e60 : SUCCESS in 10m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7328a9717d9448c3a9906afa48c57070 : SUCCESS in 10m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f847e25a79f438a8fe14c1838a3d790 : SUCCESS in 53m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f378a9aa69614ddc83c18cf64d187f3e : SUCCESS in 1h 05m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f843014e728b4fa291bcd800420c4d1b : SUCCESS in 1h 14m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/921fcf53143047bab8c457ff1a2479c7 : SUCCESS in 54m 32s","accounts_in_message":[],"_revision_number":4},{"id":"d8b6aa7b2ab8ed9cb4e47bc7aa18f82ad5a12539","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 22:41:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d0f2cb785be4ba581d67d8b603ab404 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":4},{"id":"dcb3c3c1b399146bbc73321ac45d8bf96480a382","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-18 09:08:24.000000000","message":"Patch Set 4:\n\nLooks like it is time for me to add AArch64 CI to requirements\n\nhttps://github.com/protocolbuffers/protobuf/issues/7520\n\nINFO:kolla.common.utils.openstack-base:ERROR: No matching distribution found for protobuf\u003d\u003d\u003d3.12.0 (from -c /requirements/upper-constraints.txt (line 353))","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0624d5d62edfce16c27cfc36a01e82ed08e8a3b7","revisions":{"342bdc2cd121da857d81c11108e9631bbc111e78":{"kind":"REWORK","_number":1,"created":"2020-05-15 06:16:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/728367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/728367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/728367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/728367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/728367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/728367/1"}}},"commit":{"parents":[{"commit":"84fe9d0bd2f5c78daf02ad1d1ad8bf3683b783ee","subject":"Merge \"update constraint for reno to new release 3.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/84fe9d0bd2f5c78daf02ad1d1ad8bf3683b783ee"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-15 06:16:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-15 06:16:51.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ibab788d0b31e2e1da145f20710233419b982b4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/342bdc2cd121da857d81c11108e9631bbc111e78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/342bdc2cd121da857d81c11108e9631bbc111e78"}]},"branch":"refs/heads/master"},"3fe457aefddceab853412ccf53f2209cea7a869c":{"kind":"REWORK","_number":2,"created":"2020-05-16 06:13:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/728367/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/728367/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/728367/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/728367/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/728367/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/728367/2"}}},"commit":{"parents":[{"commit":"32477163f3498c157af7e3a57e33618c117cb23b","subject":"update constraint for openstackdocstheme to new release 2.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32477163f3498c157af7e3a57e33618c117cb23b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-16 06:13:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-16 06:13:17.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ibab788d0b31e2e1da145f20710233419b982b4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fe457aefddceab853412ccf53f2209cea7a869c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fe457aefddceab853412ccf53f2209cea7a869c"}]},"branch":"refs/heads/master"},"f48018d277a2efd365ed9ca143771674799d4d5a":{"kind":"REWORK","_number":3,"created":"2020-05-16 15:57:27.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/67/728367/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/728367/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/728367/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/728367/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/728367/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/728367/3"}}},"commit":{"parents":[{"commit":"d0b389d840c75216ab2cc10df849cb98990b1503","subject":"update requirements and constraints for py3 only Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0b389d840c75216ab2cc10df849cb98990b1503"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-16 06:13:17.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-16 15:56:13.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ibab788d0b31e2e1da145f20710233419b982b4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f48018d277a2efd365ed9ca143771674799d4d5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f48018d277a2efd365ed9ca143771674799d4d5a"}]},"branch":"refs/heads/master"},"0624d5d62edfce16c27cfc36a01e82ed08e8a3b7":{"kind":"REWORK","_number":4,"created":"2020-05-16 19:40:55.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/67/728367/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/728367/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/728367/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/728367/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/728367/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/728367/4"}}},"commit":{"parents":[{"commit":"d0b389d840c75216ab2cc10df849cb98990b1503","subject":"update requirements and constraints for py3 only Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0b389d840c75216ab2cc10df849cb98990b1503"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-16 06:13:17.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-05-16 19:40:47.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ibab788d0b31e2e1da145f20710233419b982b4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0624d5d62edfce16c27cfc36a01e82ed08e8a3b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0624d5d62edfce16c27cfc36a01e82ed08e8a3b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
