)]}'
{"id":"openstack%2Frequirements~728432","triplet_id":"openstack%2Frequirements~master~I85be5656714b1cdeb6afb8ffa9f1a4124b3ebc0b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I85be5656714b1cdeb6afb8ffa9f1a4124b3ebc0b","subject":"update constraint for openstackdocstheme to new release 2.1.2","status":"MERGED","created":"2020-05-15 12:33:51.000000000","updated":"2020-05-15 23:09:19.000000000","submitted":"2020-05-15 23:07:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728432-1589584048744-0ef704bd","meta_rev_id":"de2e27e26cfe0e9a84032d3c5ec8786a6197e416","_number":728432,"virtual_id_number":728432,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-05-15 13:38:03.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-15 23:07:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-05-15 16:12:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-05-15 16:19:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-15 16:10:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-05-15 16:19:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 13:38:03.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-05-15 16:10:51.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-15 16:12:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-15 16:19:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-15 23:07:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79d3deafb89ed204e61627289e9b731463119f84","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-15 12:33:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2183dd2f87f87f6a73bd3039106f80aa5421b110","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-15 12:35:58.000000000","message":"Patch Set 1: Code-Review-1\n\nLet\u0027s give a WIP here (-1 since I can\u0027t WIP) and test cinder and nova first... I\u0027ll add some changes that depends-on this one.","accounts_in_message":[],"_revision_number":1},{"id":"1e70ca2cef16002f62f47b59c7af5d696179c1a7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-15 12:39:50.000000000","message":"Patch Set 1:\n\nTesting in:\n\nhttps://review.opendev.org/728263/\nhttps://review.opendev.org/728272\nhttps://review.opendev.org/728270","accounts_in_message":[],"_revision_number":1},{"id":"3950f4c1d4053b3e83e61f5415b1dbd2db59ced5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-15 13:38:03.000000000","message":"Patch Set 1:\n\n\u003e Testing in:\n \u003e \n \u003e https://review.opendev.org/728263/\n \u003e https://review.opendev.org/728272\n \u003e https://review.opendev.org/728270\n\nUnfortunately it would not tests as tox use released constarint not from cloned repo.","accounts_in_message":[],"_revision_number":1},{"id":"6787ed4196af37fcb752459a9c767362216aaa21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 13:51:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bd03e9df5c44754b580a6b1763c8c50 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4de4af1f779b47b3a3344d5e33323b9f : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f395331077a466fa53acdeade5f4f9b : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d97a72980914246a866cc6d40854479 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d1a0fd61d434fc992f8e0959b1d42e2 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac7e296c37c845748fe75c122db34a2e : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/29203f0fd6db4929b9052b23856576a5 : SUCCESS in 8m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9e5dca4ded7840f79d7d3846e43a714e : SUCCESS in 10m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/547aa1df7e384bf39823ca27f2b20ced : SUCCESS in 5m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5f863c4f960845f2b46656be7efeea54 : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/11a3886f84b74309866ddd6d8defc42a : SUCCESS in 9m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d1445551a54048509ba3a1bdc772daed : SUCCESS in 20m 50s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eda405d741e4428a8d4950cb324ffd95 : SUCCESS in 24m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/84d33d6075674d00b7d2e50a945c84c2 : SUCCESS in 13m 40s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6e0b63c080824f17af7ec872052cdd62 : SUCCESS in 14m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/84560b80d5694844a80b3687e03c2d05 : SUCCESS in 16m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ee66e3df3f5b48c29d6b34e579bb1349 : SUCCESS in 4m 43s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/11ab392bf7084084a1a641c01dbb736a : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/481fd05663d04a8cada1630eb580b9ea : SUCCESS in 5m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/4d3ac6636f374e1db140064aab5c6afc : SUCCESS in 12m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3863368ca66d4acc9a2da7a972d083a4 : SUCCESS in 10m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a2d317406f34bc48cafeb6b01d2c71e : SUCCESS in 51m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7fbe8deeb1d54acb9dfb8708b0640767 : SUCCESS in 1h 12m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7be90e26073a4ff79a7c802aab06bea8 : SUCCESS in 1h 13m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/9253c7c845e34b0bb0cf6ab794642c08 : SUCCESS in 1h 03m 48s","accounts_in_message":[],"_revision_number":1},{"id":"558b2457cecd2f01502e6b2094d2e4fe7587910b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-15 13:56:40.000000000","message":"Patch Set 1:\n\n\u003e \u003e Testing in:\n \u003e \u003e\n \u003e \u003e https://review.opendev.org/728263/\n \u003e \u003e https://review.opendev.org/728272\n \u003e \u003e https://review.opendev.org/728270\n \u003e \n \u003e Unfortunately it would not tests as tox use released constarint not\n \u003e from cloned repo.\n\nIt does at least partially. I\u0027m currently hunting down the bugs where it fails.\n\nIt works here:\nhttps://zuul.opendev.org/t/openstack/build/bd7ce066b0a549f683e2f04e82083598/log/job-output.txt#525\n\nBut breaks in PDF build.","accounts_in_message":[],"_revision_number":1},{"id":"689092decc7da70a8b81b241cea516ee66449fa1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-15 15:26:50.000000000","message":"Patch Set 1:\n\nok, does this need to be blacklisted as well?","accounts_in_message":[],"_revision_number":1},{"id":"ba171e026e34d309b21bbb07523ddf8ed3247fa5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-15 15:30:43.000000000","message":"Patch Set 1:\n\nNo need to black list this, we fixed zuul-jobs and openstack-zuul-jobs bugs that our testing triggered - and now python-novaclient is building correctly with a depends-on this change (failed before).\n\nDepends-On is now working correctly for constraints also for PDF.\n\nLet\u0027s wait for the results from nova and cinder whether those are fine as well.\n\nSee https://review.opendev.org/#/q/status:open++topic:reno-openstacksdocstheme for the changes I have up.\n\nLet\u0027s keep my -1 as WIP, I\u0027ll change to +1 once tests ar efine.","accounts_in_message":[],"_revision_number":1},{"id":"004aa45409b61b5d96de54f1891d95943f7328e4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-15 16:10:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"698144a4426305ec5e0707e29da14e9b62695759","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-15 16:11:45.000000000","message":"Patch Set 1:\n\nnova and novaclient are green, cinder and oslo.concurrency are passing docs jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad0368a9ad5ba0be80bfff2d25f095a7bcc92130","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-15 16:12:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0f8d71fde2963e8d00049ab3b7fce783db01bd2a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-15 16:19:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d763d7c7bb6a0051daf62ca11954261c28b401b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 16:20:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae7107e72771b40bd872960237ddf13a059d46f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 17:33:40.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"82f393bb1368f02ce1e4a9badba7f1009857ab39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 23:07:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a38be62e960181afa0a23f4ad439ca156de21b29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 23:07:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3bdfcf92ea6453ca0a04ded2b241e9b : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb7bd0eea2304116937581981641c296 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d178e154b524f05a6e4b22e45fc057f : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc18dbc189b5491cbcc362788fdc272a : SUCCESS in 7m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/519cb2138fc74b48bb3f627b8a1d2f16 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/99142ee53a9a4cf9902c9302fc761ee3 : SUCCESS in 6m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8d9f6ad9b9d04799bf0f7aeddf7a2b52 : SUCCESS in 11m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ebcbdcddabe14203ae5b08699e2df747 : SUCCESS in 11m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f92e4e6ad3934163ba68445a2ae803a6 : SUCCESS in 8m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/81db05021cc94987a6832e092fb8597d : SUCCESS in 12m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29dc2ff850fb44e895fd8c0252a9fc60 : SUCCESS in 7m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b3416ba080484e3e85205ae092416169 : SUCCESS in 15m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3fb7190554b04a23a945b6b052498ec5 : SUCCESS in 28m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f86760907b9e420a9070939df84a9f61 : SUCCESS in 15m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/610a5079c74648ba8d91e6e14ceb2058 : SUCCESS in 14m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/55dfb858463b4c38809b6c1255235386 : SUCCESS in 19m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e701e64498bf4d17b414dd579292c9f2 : SUCCESS in 5m 44s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9ef4e5d6346844f6b0fde6419e6c1152 : SUCCESS in 10m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0511a6dfaa854af4a92bfbef5858f722 : SUCCESS in 10m 05s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2df4bbaf73cf4075bac5421dd11133f4 : SUCCESS in 16m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/98af1d3630d44a649bb7105e8e8e273b : SUCCESS in 15m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7223ff4a4ad4a38adbdc3896b1dbee0 : SUCCESS in 1h 09m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f79f1bcda0d14deeadabd1373ab17e6f : SUCCESS in 1h 16m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d557f54efc5746bf9e34152c0a7e98f4 : SUCCESS in 1h 18m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/d0fc131937ea4aba801961aadd3b2522 : SUCCESS in 1h 01m 50s","accounts_in_message":[],"_revision_number":1},{"id":"985be26c2506a84c9d7f6a35903b39b2d36da3b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 23:09:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbb8ce1c359143b89c5a78795bfbed51 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32477163f3498c157af7e3a57e33618c117cb23b","revisions":{"32477163f3498c157af7e3a57e33618c117cb23b":{"kind":"REWORK","_number":1,"created":"2020-05-15 12:33:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/728432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/728432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/728432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/728432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/728432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/728432/1"}}},"commit":{"parents":[{"commit":"a5cfd10b8c6775b4630f9693f1d5177588137103","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5cfd10b8c6775b4630f9693f1d5177588137103"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-15 12:33:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-15 12:33:49.000000000","tz":0},"subject":"update constraint for openstackdocstheme to new release 2.1.2","message":"update constraint for openstackdocstheme to new release 2.1.2\n\nChange-Id: I85be5656714b1cdeb6afb8ffa9f1a4124b3ebc0b\nmeta:version: 2.1.2\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Andreas Jaeger \u003caj@suse.com\u003e\nmeta:release:Commit: Andreas Jaeger \u003caj@suse.com\u003e\nmeta:release:Change-Id: Id68466160711a7c77ccd6baeefe47896d0a8ca48\nmeta:release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32477163f3498c157af7e3a57e33618c117cb23b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32477163f3498c157af7e3a57e33618c117cb23b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
