)]}'
{"id":"openstack%2Frequirements~729306","triplet_id":"openstack%2Frequirements~master~I4e1d6ea12682367f43fcdd0414327a9d68d8adc7","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I4e1d6ea12682367f43fcdd0414327a9d68d8adc7","subject":"Cap zipp for py3.5","status":"MERGED","created":"2020-05-19 15:48:04.000000000","updated":"2020-05-21 21:49:33.000000000","submitted":"2020-05-21 21:48:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729306-1590097686528-4cad9dd9","meta_rev_id":"15ff16e3d65f77dcc5de41990b2f93c306845618","_number":729306,"virtual_id_number":729306,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"date":"2020-05-21 16:05:55.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-21 21:48:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-05-21 16:00:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-05-21 15:59:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-05-21 16:00:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 18:02:22.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2020-05-21 15:59:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-21 16:00:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-21 21:48:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dafbd31181c52aabcb8c66f3f9973bf8af583008","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-19 15:48:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1269f03b989551d7021a26803fbd02750f5dc5d7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-19 15:49:28.000000000","message":"Patch Set 1: Code-Review-1\n\nwhere are we using py35 in openstack (master), iirc victoria was py36/38 only","accounts_in_message":[],"_revision_number":1},{"id":"0e60401fbcbf9b731e0aa44bc0d49a3f8ee0979b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-19 15:53:35.000000000","message":"Patch Set 1: Code-Review-1\n\nYeah, we only support py36 or later now, so this shouldn\u0027t be necessary.\n\nAny repos that want to continue to test against unsupported versions need to maintain their own constraints list.","accounts_in_message":[],"_revision_number":1},{"id":"24f503c39842e556ef92002a0b1a84cb5d8d2356","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-19 16:06:18.000000000","message":"Patch Set 1:\n\nas commented inline few projects still support the py3.5 on master. let me add doc link also.","accounts_in_message":[],"_revision_number":1},{"id":"f478313bf6e56fec86d5ef2e3fd04d78ae1f4629","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-19 16:06:44.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b8960f5316af88322aa7a342f386306a7c535682","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-19 18:02:00.000000000","message":"Patch Set 2:\n\nhacking gate is back with this - https://review.opendev.org/#/c/729317/.","accounts_in_message":[],"_revision_number":2},{"id":"d5a54d75232d9b1b970ffc112a402da5e878b85c","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-05-19 18:02:22.000000000","message":"Patch Set 2:\n\n\u003e where are we using py35 in openstack (master), iirc victoria was\n \u003e py36/38 only\n\nThere are some exceptions: \n\nhttps://review.opendev.org/#/c/721084/","accounts_in_message":[],"_revision_number":2},{"id":"a36634f06d39aeb8ac683a5171d8c819778832b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 18:09:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a84e3f6e740249b1aa6d710fea3db44b : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02b66f5275884800b29e53cc76231e9e : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a47247911f2467fb9b541dff6d6fe4b : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32935598a520423ba4bc2bbdb472d29d : SUCCESS in 5m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e04de918981b4b8c8f41fc78d71ff82e : SUCCESS in 6m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8116391151154808bb12d22521af3b5d : SUCCESS in 6m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ee7946319ad0495c88cd367bf189fe05 : SUCCESS in 10m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/02a9e9198e13450db0760ee91fd37615 : SUCCESS in 10m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/175c0585b61d420aa49eceeb2eb37b39 : SUCCESS in 11m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2faf378efc98411aa6351548dfa7b480 : SUCCESS in 8m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/eaba10e7b01949e4b58d8762389dcfb6 : SUCCESS in 12m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/44222991310e48d4b2e999cd2f31969e : SUCCESS in 10m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cafea34b14e844f387e44b505b09a486 : SUCCESS in 15m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/32cbcdb826f04a8498bc9057933f7e86 : SUCCESS in 28m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0c65fceb4e224d8fad9749d8a1127262 : SUCCESS in 12m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3496469e6fe34c5cb9dd3501a54b604b : SUCCESS in 19m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/07c7d0732bd2419b809b4b94d7ec06f3 : SUCCESS in 7m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/48e8701611e64017b8fa20fb6501c959 : SUCCESS in 8m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0b32bfb4f721462e9af6dcf3ef2536c4 : SUCCESS in 14m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3eaef546d84f48b889175dcfb93bc8d2 : SUCCESS in 51m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/200b1f95a67e4d7da375c34c84329d64 : SUCCESS in 1h 20m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a04be96065104050b7ecb378cc220832 : SUCCESS in 1h 29m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/7b2cd918b65b4f268682c4d272c963f8 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d850443f06b796ec4bd02c21c6bb8ef0cb645f68","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-19 20:34:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"05e9a72007548818fedd10db2be340191b7d4ed0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-19 23:20:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6e9b1906bc2bb1ccee4bdff6e39edfe7483385d6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-20 21:43:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c6a0350367f63561bb5969bdaa2c28d70656271","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-21 01:30:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"81c9801e467b1cce2b0bfafb4157fa30def9a625","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-21 01:46:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7b3e8f8bf15e23cfc7f1fc9707f9fe021e5d956","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-21 02:40:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d5d8c5a9c3c50faceebbe825fd1a5c017bfe1072","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-21 02:41:26.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"30a17ce491fe10b08263d98d6ed5fa146f8f142e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 04:20:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3365244768a84a59968f451681b9e5a5 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f9a76071f27497cbab88559e80efb77 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb91a7fdc7694785a534499bdcafc1b2 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/640ea3384431403abc557184acc0d73c : SUCCESS in 4m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f0fc50443b224ff2b24492c6e8770e8a : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d2d4bae2db454d848f742f58feb39ac4 : SUCCESS in 4m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6db392e8b9434846a620bf6e8f732bae : SUCCESS in 8m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/033f293f9d98410da5d92d922a94d6fc : SUCCESS in 9m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5dbd6653b9324e7280dd65e91c849fb7 : SUCCESS in 9m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e8df1887771245349d00c27efc50110c : SUCCESS in 6m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f089abc17cee4067b7ab302513c2cffc : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/753ede787b8649398d8e351030a56875 : SUCCESS in 8m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/db255f6c4a3e47b4928cb62aa204f9d2 : SUCCESS in 12m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/53492c0e067345d1abbb94a51a5f746f : SUCCESS in 36m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5d81a45b403b483992ce5d046e4bf625 : SUCCESS in 16m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a24dac66bdf441d0967c1dd4329ee714 : SUCCESS in 16m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/15122d333d104a9fbc17583d48aad17a : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/82e0c0a4b2ad4247a53dab1b8ff702a5 : SUCCESS in 7m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/56210a9457de464fa8b38d3f0a4b6c36 : SUCCESS in 11m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1dffb52a549c40faaa9c9906f3003a18 : SUCCESS in 1h 01m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8625665e7374cc7813224b20a37fd4a : SUCCESS in 55m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df382229012946cda3453495f769d535 : SUCCESS in 1h 16m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/6eecfafd4c404f2c9088f5315abd6f34 : SUCCESS in 1h 38m 28s","accounts_in_message":[],"_revision_number":3},{"id":"6f3bae1888d41648d3a694bbeffd3bc310750868","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-21 05:16:00.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"660ef162bc2fa567731899a75198e20410908ddc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-21 12:11:24.000000000","message":"Patch Set 3: Code-Review-1\n\nThis is missing 3.7 and 3.8 upper constraints still.","accounts_in_message":[],"_revision_number":3},{"id":"14616e84962fc1eebee94b2bd38c7a7881d2c6af","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-21 15:43:40.000000000","message":"Patch Set 3:\n\n\u003e This is missing 3.7 and 3.8 upper constraints still.\n\npython_version\u003e\u003d\u00273.6\u0027 will not work?","accounts_in_message":[],"_revision_number":3},{"id":"51732b0a3c7c49e0c242fa5e34949e86574f1b67","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-21 15:55:31.000000000","message":"Patch Set 3:\n\nya, I think you are fine, it\u0027s \u003e\u003d, though I suppose the format preferrs \u003d\u003d for each python revision instead of  \u003e\u003d\n\nfoo.bar\u003d\u003d\u003d0.1.2;python_version\u003d\u003d\u00273.6\u0027\nfoo.bar\u003d\u003d\u003d1.2.3;python_version\u003d\u003d\u00273.7\u0027\nfoo.bar\u003d\u003d\u003d1.2.3;python_version\u003d\u003d\u00273.8\u0027","accounts_in_message":[],"_revision_number":3},{"id":"e0f826d1016ac33f13b42a0a4a564ae3aaf85093","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-21 15:57:56.000000000","message":"Patch Set 3:\n\n\u003e ya, I think you are fine, it\u0027s \u003e\u003d, though I suppose the format\n \u003e preferrs \u003d\u003d for each python revision instead of  \u003e\u003d\n \u003e \n \u003e foo.bar\u003d\u003d\u003d0.1.2;python_version\u003d\u003d\u00273.6\u0027\n \u003e foo.bar\u003d\u003d\u003d1.2.3;python_version\u003d\u003d\u00273.7\u0027\n \u003e foo.bar\u003d\u003d\u003d1.2.3;python_version\u003d\u003d\u00273.8\u0027\n\nyeah, both should work but let me update it to be consistent with what we used to have.","accounts_in_message":[],"_revision_number":3},{"id":"75e767f7510fe89d13e677ad61f0dfca850a79e2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-21 15:58:44.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c469ee491340bbd1cb38befeee14c469e62b5a3b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-21 15:58:55.000000000","message":"Patch Set 3:\n\n\u003e \u003e ya, I think you are fine, it\u0027s \u003e\u003d, though I suppose the format\n \u003e \u003e preferrs \u003d\u003d for each python revision instead of  \u003e\u003d\n \u003e \u003e\n \u003e \u003e foo.bar\u003d\u003d\u003d0.1.2;python_version\u003d\u003d\u00273.6\u0027\n \u003e \u003e foo.bar\u003d\u003d\u003d1.2.3;python_version\u003d\u003d\u00273.7\u0027\n \u003e \u003e foo.bar\u003d\u003d\u003d1.2.3;python_version\u003d\u003d\u00273.8\u0027\n \u003e \n \u003e yeah, both should work but let me update it to be consistent with\n \u003e what we used to have.\n\nNo, this will not work. It needs to be separate like:\n\nhttps://opendev.org/openstack/requirements/src/branch/stable/ussuri/upper-constraints.txt#L940-L943\n\nOtherwise it will not get updates:\n\nhttps://opendev.org/openstack/project-config/src/branch/master/roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh#L124","accounts_in_message":[],"_revision_number":3},{"id":"6ab0c6ad261811af43a08c8a76d4669c6f3c6196","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-21 15:59:08.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f9ddb919e983e6243137502665183e90ed420f5c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-21 16:00:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b531c5d3dfe6ca5bb7a8e13d867cfc3bc3b8f8a3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-21 16:05:55.000000000","message":"Patch Set 4:\n\n\u003e \u003e \u003e ya, I think you are fine, it\u0027s \u003e\u003d, though I suppose the format\n \u003e \u003e \u003e preferrs \u003d\u003d for each python revision instead of  \u003e\u003d\n \u003e \u003e \u003e\n \u003e \u003e \u003e foo.bar\u003d\u003d\u003d0.1.2;python_version\u003d\u003d\u00273.6\u0027\n \u003e \u003e \u003e foo.bar\u003d\u003d\u003d1.2.3;python_version\u003d\u003d\u00273.7\u0027\n \u003e \u003e \u003e foo.bar\u003d\u003d\u003d1.2.3;python_version\u003d\u003d\u00273.8\u0027\n \u003e \u003e\n \u003e \u003e yeah, both should work but let me update it to be consistent with\n \u003e \u003e what we used to have.\n \u003e \n \u003e No, this will not work. It needs to be separate like:\n \u003e \n \u003e https://opendev.org/openstack/requirements/src/branch/stable/ussuri/upper-constraints.txt#L940-L943\n \u003e \n \u003e Otherwise it will not get updates:\n \u003e \n \u003e https://opendev.org/openstack/project-config/src/branch/master/roles/copy-release-tools-scripts/files/release-tools/update_constraints.sh#L124\n\nah yes, updating will be the issue in that case. thanks.","accounts_in_message":[],"_revision_number":4},{"id":"714624c86a3646505d2a6b924bd9fd39d96e9727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 17:32:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f8512c6c54e404d8d4e11559411069e : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1a2ddd096d74db680dd31d58b85c218 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a23548a0c3624cb28b4df356597ac1ee : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5336a1108e634ad383fe1a64b88e6803 : SUCCESS in 6m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9dacad2a80f42e3b69f7f4822dad80f : SUCCESS in 7m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd3682e3f3f94cff82b559899a700142 : SUCCESS in 7m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/134b14e9cc1f4819a3df0ac425d42659 : SUCCESS in 11m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/458fb7dec8234f1d83b40ee4bf7ed1e1 : SUCCESS in 10m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a33541ba10ca40ff894361ad78d1b2b7 : SUCCESS in 12m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0e315f9548b24e19a2107ba33498951b : SUCCESS in 5m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/144688c20342471b81b9f2e6787383b0 : SUCCESS in 12m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/78ad997195744ccead990834fe24a8a4 : SUCCESS in 9m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/000f9c6b7ff3402b96ffde448f745551 : SUCCESS in 15m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/554350ffc1994d0bb6f939264830ef7e : SUCCESS in 31m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e8abf70cd2884e0d9c1742cef654a256 : SUCCESS in 20m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/69e15b1fa4db4c348823a5a7a050ebe9 : SUCCESS in 19m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/634c1434fec14ee7a3b562b036dfbaa5 : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aa2b4953a4334f63bba698a179b4dbea : SUCCESS in 7m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/83aa9fdd9ce943fdbccf571b612de76d : SUCCESS in 10m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9a39c75ccef4aabae02990064837cb9 : SUCCESS in 57m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1db89290dba4031881a957d0a50651e : SUCCESS in 1h 14m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37a29eddc9bb443f8380467f8752ac30 : SUCCESS in 1h 28m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/c7bf74757a5345628e47b79f7ef82c66 : SUCCESS in 1h 04m 15s","accounts_in_message":[],"_revision_number":4},{"id":"e67c8343d5943401f3681bc2c790a11ef869a9a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 17:32:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c25a8208005a869e69cb35b5ff87dc99ea878173","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 21:48:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c7d2f98f963e3cf3a56d6cf79e3fe7e9962d8bc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 21:48:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fefff92fa1f94791a06b5919ac202dc5 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c31a2aa2a544b91bd83dfba66de95d0 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf456d43fab541418bb926116016e844 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11372a53e6a3461ea5d228860d95ba1e : SUCCESS in 5m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/055aedc791c040b1882c1226f396226e : SUCCESS in 6m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f0adc4d47ba44cdd9d447a4de6d73323 : SUCCESS in 6m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/46fcadc60c044f939f6b0142634cb805 : SUCCESS in 9m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/00b902606c9449f7a5ee0bd2c34b0fc4 : SUCCESS in 13m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fbc43054fcf7430290e84722ea24f39f : SUCCESS in 10m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4f605c7bc74b4f8d90dd8359af14e37a : SUCCESS in 7m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3e6060e8b8a34d1fa960e46039ba17f6 : SUCCESS in 12m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d7c23863c1ae4aa88114e1250eda7a4c : SUCCESS in 7m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1c31e8d8dd2b4ec2bfa21570aad71b04 : SUCCESS in 30m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ee77f4b38da242278f26de49a4cf657b : SUCCESS in 26m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a2b328ab4e9a45b5a450c5640261ec90 : SUCCESS in 14m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4e251e254f8a49e2b933966d25f34318 : SUCCESS in 20m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/70b6bc8f57e84d47ba71c460f213d452 : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d937641482a84b1f930b9f19bf1005de : SUCCESS in 7m 42s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/9d72e20aae3a4148bc2d3fee8352e223 : SUCCESS in 15m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1e9ef25ad5fb4a6e9f567f15dd6da940 : SUCCESS in 15m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e157f6b977a5454388e752c6c3766c62 : SUCCESS in 52m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/99f698d583d94839bff219de3ae0ddc7 : SUCCESS in 1h 04m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/437503c750cb411ab50a3fd715e308e3 : SUCCESS in 1h 21m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/2457d3e633bc4cadaa2e466d4200e449 : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":4},{"id":"15ff16e3d65f77dcc5de41990b2f93c306845618","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 21:49:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e34d34216154e0f9cf7784a8c72435f : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b280662b9dfa551b93d372336672fd52cdb71638","revisions":{"e9084de830293cd1f770c7e28885ca0d0a92205d":{"kind":"REWORK","_number":1,"created":"2020-05-19 15:48:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/06/729306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/729306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/729306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/729306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/729306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/729306/1"}}},"commit":{"parents":[{"commit":"1b8b07f56186207044d379b72867b7d9ea7c66b0","subject":"Merge \"update constraint for openstackdocstheme to new release 2.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b8b07f56186207044d379b72867b7d9ea7c66b0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-19 15:46:39.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-19 15:46:39.000000000","tz":-300},"subject":"Cap zipp for py3.5","message":"Cap zipp for py3.5\n\nWe fixed it many time for example\n- https://review.opendev.org/#/c/705184/3\n\nBut with latest req update fro Victoria uncapped it and caused\nfailure on py3.5 jobs on projects still supporting the py3.5\nlike hacking, stackviz etc.\n\nChange-Id: I4e1d6ea12682367f43fcdd0414327a9d68d8adc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9084de830293cd1f770c7e28885ca0d0a92205d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9084de830293cd1f770c7e28885ca0d0a92205d"}]},"branch":"refs/heads/master"},"59043de408f24462dc6df2e28e0b52b5412b2643":{"kind":"REWORK","_number":2,"created":"2020-05-19 16:06:44.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/06/729306/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/729306/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/729306/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/729306/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/729306/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/729306/2"}}},"commit":{"parents":[{"commit":"1b8b07f56186207044d379b72867b7d9ea7c66b0","subject":"Merge \"update constraint for openstackdocstheme to new release 2.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b8b07f56186207044d379b72867b7d9ea7c66b0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-19 15:46:39.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-19 16:06:42.000000000","tz":0},"subject":"Cap zipp for py3.5","message":"Cap zipp for py3.5\n\nWe fixed it many time for example\n- https://review.opendev.org/#/c/705184/3\n\nBut with latest req update fro Victoria uncapped it and caused\nfailure on py3.5 jobs on projects still supporting the py3.5\nlike hacking, stackviz etc.\n\nChange-Id: I4e1d6ea12682367f43fcdd0414327a9d68d8adc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59043de408f24462dc6df2e28e0b52b5412b2643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59043de408f24462dc6df2e28e0b52b5412b2643"}]},"branch":"refs/heads/master"},"1c7335ecc27fc770f90b1981ea429ccfd51ec706":{"kind":"REWORK","_number":3,"created":"2020-05-21 02:41:26.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/06/729306/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/729306/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/729306/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/729306/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/729306/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/729306/3"}}},"commit":{"parents":[{"commit":"1b8b07f56186207044d379b72867b7d9ea7c66b0","subject":"Merge \"update constraint for openstackdocstheme to new release 2.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b8b07f56186207044d379b72867b7d9ea7c66b0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-19 15:46:39.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-21 02:41:22.000000000","tz":0},"subject":"Cap zipp for py3.5","message":"Cap zipp for py3.5\n\nWe fixed it many time for example\n- https://review.opendev.org/#/c/705184/3\n\nBut with latest req update fro Victoria uncapped it and caused\nfailure on py3.5 jobs on projects still supporting the py3.5\nlike hacking, stackviz etc.\n\nChange-Id: I4e1d6ea12682367f43fcdd0414327a9d68d8adc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c7335ecc27fc770f90b1981ea429ccfd51ec706"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c7335ecc27fc770f90b1981ea429ccfd51ec706"}]},"branch":"refs/heads/master"},"b280662b9dfa551b93d372336672fd52cdb71638":{"kind":"REWORK","_number":4,"created":"2020-05-21 15:58:44.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/06/729306/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/729306/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/729306/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/729306/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/729306/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/729306/4"}}},"commit":{"parents":[{"commit":"1b8b07f56186207044d379b72867b7d9ea7c66b0","subject":"Merge \"update constraint for openstackdocstheme to new release 2.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b8b07f56186207044d379b72867b7d9ea7c66b0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-19 15:46:39.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-21 15:58:42.000000000","tz":0},"subject":"Cap zipp for py3.5","message":"Cap zipp for py3.5\n\nWe fixed it many time for example\n- https://review.opendev.org/#/c/705184/3\n\nBut with latest req update fro Victoria uncapped it and caused\nfailure on py3.5 jobs on projects still supporting the py3.5\nlike hacking, stackviz etc.\n\nChange-Id: I4e1d6ea12682367f43fcdd0414327a9d68d8adc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b280662b9dfa551b93d372336672fd52cdb71638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b280662b9dfa551b93d372336672fd52cdb71638"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
