)]}'
{"id":"openstack%2Frequirements~729957","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Ifb78827fc967cb91e3cb7b9816566dc1b1fff4f9","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Ifb78827fc967cb91e3cb7b9816566dc1b1fff4f9","subject":"update constraint for ironic-lib to new release 2.21.1","status":"MERGED","created":"2020-05-21 13:42:57.000000000","updated":"2020-05-21 22:32:36.000000000","submitted":"2020-05-21 22:32:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729957-1590100356113-fdce670a","meta_rev_id":"f16e4465e40fd36b7e3822fee29bd8e5096d42a3","_number":729957,"virtual_id_number":729957,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-21 22:32:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-21 13:44:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-21 20:11:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-21 20:11:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 13:44:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-21 20:11:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-21 22:32:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d89a7724939df379391b5025c551ebb9f0eb0070","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-21 13:42:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fd120c05d56dcbe5018de9c288724410b0ab4f3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-21 13:44:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"30c12bc36cdb1d53d2eabdc0d5dba37cdcf7fa11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 15:56:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b28319b81f2e44e088f1f7a77d31e505 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5954db936e1a48ddbd6adff3dac4b7a9 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6df19fc67d9d4feabc27183292e32b22 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2e4af5cd254486d9fb02040316269a7 : SUCCESS in 4m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a795689e8c3145dd81092eff281761c2 : SUCCESS in 6m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e43f422529f24bd183c73ff36009bc1c : SUCCESS in 5m 13s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/4562573d0c3d440f9c286499ae71cddf : SUCCESS in 8m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/244eed8d53cd43aea6cded2553b5d69c : POST_FAILURE in 8m 26s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2e61acacac004155bfc633f4e9ca1a9d : SUCCESS in 10m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/850644c00dac425da03093982fa39ada : SUCCESS in 12m 11s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/506cdafedaad4c84a07bb3f8bb485246 : SUCCESS in 6m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eb3403c4653248b79a864d198a05f9da : SUCCESS in 5m 40s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3b95b834b9ea4cc3b219d31c7a9d6a3e : SUCCESS in 14m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2c739983856140c38c6095e85003acd0 : SUCCESS in 12m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b550e5237fe0493193e5eca7038126d8 : SUCCESS in 12m 32s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2cc3c90d039d4294a4bc56e1b493699a : SUCCESS in 24m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/74449ad4db6f48d2a488ea3e8e8c21a1 : SUCCESS in 27m 08s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/599d1a1bdaea4eec9b47b1af5ed0fdff : SUCCESS in 23m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/48df05f3bf954838b8f1d9681511bab9 : SUCCESS in 27m 53s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f38994bcb62a47a0a8bf12808f1520ef : SUCCESS in 18m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/aafc86c4940c4ee7a145475d9ebfed59 : SUCCESS in 15m 35s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a51142242876414ca1dbe9972a23d0e9 : SUCCESS in 24m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2c53a82a00f548d5b2c331812984947d : SUCCESS in 22m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e810f045c40f40bfa483b174665391eb : SUCCESS in 6m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/670a4a4992c946709f69674f668677cb : SUCCESS in 10m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b185568041a94bfe9589f5500beaa60d : SUCCESS in 10m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c9239563e084d429468b479c5d55b84 : SUCCESS in 1h 04m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5adca700fd6c46769aad1287ca42db00 : SUCCESS in 1h 53m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/58b8906ee0fe47bdac89a36e61b1930f : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":1},{"id":"42727f4a075466a85824eedce3416cb71232e33e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-21 16:01:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1f43665fefb354cf4b75b31e5116b650e85bf86a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 17:51:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3b314f5bf2c44a7b151865d81fcfeae : SUCCESS in 6m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a204d0aa28ff42eb9e1de40f46fcbccc : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d71110652bed417ea5ad6afb49cd5c96 : SUCCESS in 9m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2244d758c2ef41a593d77b97c88b67e0 : SUCCESS in 10m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/16dfc09b5ec74c31a8d42109587c3490 : SUCCESS in 8m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fb23af4ea2e04d1495d333609af732ff : SUCCESS in 9m 48s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8a0260d699db44a2a457c8868241175a : SUCCESS in 12m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b78be685682d4578ab05aa74a8d2424c : SUCCESS in 11m 19s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/87a7098c30ed43b595a1ac1c820bafee : SUCCESS in 16m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f00a35605bbc476da1e0b1b341cb6034 : SUCCESS in 14m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/8245d63b63b841e5be50b354b6e7ae53 : SUCCESS in 9m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ee42916c34d64664b092b684a713d65d : SUCCESS in 7m 34s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ff96f5674d084d2dbf8d210576d53da4 : SUCCESS in 11m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/eaf8a66668a44b33aeb8e932407fa0a3 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8464e40c92ad4592997db13b842f6a3f : SUCCESS in 14m 47s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a536ef4719a545d0865f4153ad5e8b09 : SUCCESS in 16m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/367bef2d20284a56a95630e135b9773b : SUCCESS in 15m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d24c0aad480d4598b3f1bcf1064c6941 : SUCCESS in 23m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/70eb03e550334d95a5656336afaa3d8a : SUCCESS in 21m 46s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/db72c143a9044461ab5aebdd08e9f60f : SUCCESS in 17m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/81c50946b5444665b20590ff8d2f4e9c : SUCCESS in 14m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f33bdf9ff4b24473add2c9d8110f5113 : SUCCESS in 17m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce04f6d5be8a4e12a5fd795196691c13 : SUCCESS in 20m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/68162df4a3424b21a05b6dc7e33267b2 : SUCCESS in 6m 09s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a45974e7ee6d477db7ddcf8311e6fbb3 : SUCCESS in 15m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f7b53e2e2a19461c8bfc3b54edb402b2 : SUCCESS in 10m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e8e582f66ff34f548e23e769f1472081 : SUCCESS in 1h 40m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/477619f2a01741789bbb49a934170b5e : SUCCESS in 1h 32m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/ae3b70e323094c0c9586a0df7e4892f9 : SUCCESS in 1h 02m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0e1ead064d870d3deb3025a9d19d005bef20bef3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-21 20:11:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e4e30f7c456dd78b6366a9413656bfbd9b4daed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 20:12:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b3721db95f8c771970e5712798dfb4a8567115e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 22:32:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e336f05491aa438dbb704ea3203fe814 : SUCCESS in 6m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ec88ee56bccc4890a90072e419f49456 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08cd784c1127429790b078df24d56471 : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/140f46ce0c7e47adad44f50d600514b2 : SUCCESS in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/860b10f21fc4442cbf1c5e4d0ee24ac8 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/04ba16db55a346c2a6efe9352af7794b : SUCCESS in 3m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a368fdebc9774b63bcd31272a001a636 : SUCCESS in 9m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fb0c151b25ea4107bf0f54a1b0840090 : SUCCESS in 8m 25s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1c4aada09cb245d4bcef3f91168ca8a0 : SUCCESS in 11m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fd6df2318501467f9ec5bb7f549b341d : SUCCESS in 11m 08s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/bea1bb34694c47a386e568d3ea27ac24 : SUCCESS in 7m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1379c922f7934ff0bb779aa0f89a30e7 : SUCCESS in 6m 19s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/136cdea2c1854a6ea3b37d855da45b46 : SUCCESS in 11m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/088aa62595ef43a08521ec565877261b : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df6e834445e04fd28ed4f97fa460c985 : SUCCESS in 10m 06s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/ccae81f7aa87405695dd6f1250eaa891 : SUCCESS in 14m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3010d8019b3e4694a7f4799ae800792d : SUCCESS in 26m 05s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1a5c65c0e03f4101913bfe75fe0982d6 : SUCCESS in 24m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e7f7a5b2337d4a7892ca03792bd43fc0 : SUCCESS in 22m 15s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/746003f8eb7f4cada5a6c5c72f17fb79 : SUCCESS in 13m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4660fd5f5a1146878b95e7418c9c69b4 : SUCCESS in 17m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/50afb43bbd984bbebe24d1d9ae11fe2e : SUCCESS in 14m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7a7d4e07b73a4b018963e7ab5bc91ea7 : SUCCESS in 20m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b084f28181f84dd3bd2dd81422bcd9b0 : SUCCESS in 5m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/30b93e57eae74302af2e91a21eb7fa4c : SUCCESS in 13m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/51d12dc432d14de08c409b5dd971d881 : SUCCESS in 15m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/38038d853c2248128c8da9a09039a717 : SUCCESS in 1h 04m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/aef533b21dd84d0b909a84dc060ed423 : SUCCESS in 1h 25m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6691b21b5b304e6585ced79566780510 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":1},{"id":"18c02a739a3064796376fc474596c4e71e53f3e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 22:32:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83272b2e421b7a023f357bfd8375fc08e8a678df","revisions":{"83272b2e421b7a023f357bfd8375fc08e8a678df":{"kind":"REWORK","_number":1,"created":"2020-05-21 13:42:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/729957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/729957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/729957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/729957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/729957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/729957/1"}}},"commit":{"parents":[{"commit":"b2c5713f7ffa23a98465d224e5d8b596e3ea11a6","subject":"Merge \"Stop testing with pypy\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2c5713f7ffa23a98465d224e5d8b596e3ea11a6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-21 13:42:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-21 13:42:55.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 2.21.1","message":"update constraint for ironic-lib to new release 2.21.1\n\nChange-Id: Ifb78827fc967cb91e3cb7b9816566dc1b1fff4f9\nmeta:version: 2.21.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Commit: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Change-Id: I0c26817ffe901389c2abedf647fe8f3230307861\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+1: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83272b2e421b7a023f357bfd8375fc08e8a678df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/83272b2e421b7a023f357bfd8375fc08e8a678df"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
