)]}'
{"id":"openstack%2Frequirements~729964","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I8643c46c338560ff8293db504691ea8e38d4101b","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I8643c46c338560ff8293db504691ea8e38d4101b","subject":"update constraint for sushy to new release 2.0.3","status":"MERGED","created":"2020-05-21 13:50:28.000000000","updated":"2020-05-21 21:48:05.000000000","submitted":"2020-05-21 21:48:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729964-1590097685370-fdce670a","meta_rev_id":"e0f6675958f23797fd7113e56779f7e82160f06c","_number":729964,"virtual_id_number":729964,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-21 21:48:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-21 14:01:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-21 16:01:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-21 16:01:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 14:01:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-21 16:01:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-21 21:48:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce5371b70c7d14b4a910cd19f43548de63699d5b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-21 13:50:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50485853c8e887c23609a22ab398914b3706bf87","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-21 14:01:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b928519c79078dc90a44a28fd513d4fae8ea1ab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 15:51:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5efcc891d2674b4c9860c263e326aa6c : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c19f8c8e66cc41ed8723b6e3994aa98b : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/032b9295cf3d4874911a918b1777dfd4 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c5aa1d385004a4abf38b7d6c7fce7f8 : SUCCESS in 8m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b3d1b9b7f9944bb98b6c6c8a2f827550 : SUCCESS in 6m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c3957e74989b441a9dceab2497169bf8 : SUCCESS in 5m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/1fe9d223a32d4a2fb6a69585f9ae2e3f : SUCCESS in 11m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3be9ca7105c74ce0b1b1665e83f28628 : SUCCESS in 11m 11s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/217c14bfa94142f28f40b34b9a9969e7 : SUCCESS in 11m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8ef654a39d1043a48508d0c393bc1e89 : SUCCESS in 13m 23s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/259bbbe252924bee8ca30397c89ad0ae : SUCCESS in 6m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/86f3820379344223aac0f70ee0856336 : SUCCESS in 8m 03s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/e67e38c397ae42a6be8ec962027e2057 : SUCCESS in 14m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f5ef97d599074a3897d5329162b63f4f : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/981754eaee5b47ec8db003fba1519073 : SUCCESS in 10m 10s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4f7fea0d7d9c45d2ae20f7b9873aa976 : SUCCESS in 21m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fb003f2928524c51a5573e1da22653a2 : SUCCESS in 16m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1c6501e53c384d34998ba685e0518d65 : SUCCESS in 31m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5c856f11aecf455888f6d2f5bf1b9977 : SUCCESS in 34m 48s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/da4354fc88d64b6c91e2ad2150f5d31a : SUCCESS in 17m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/903e059b0bb34069aef3dab68c09861c : SUCCESS in 14m 26s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c6b231a774d142efa8b6c77dfc2afa5c : SUCCESS in 21m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2d62182df5764b78b41d0b526ac7263b : SUCCESS in 20m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d3dc2677a70541349ec8532d64b7f55b : SUCCESS in 7m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c0e7b8feda9d4acabc049fea6facf35b : SUCCESS in 12m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ee3c974ab1d440ed9c721bf950ce9eb8 : SUCCESS in 16m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/94ed8c0eff3f499098db4bd61218185c : SUCCESS in 1h 09m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2ec461f0094547978d2c87312418afbf : SUCCESS in 1h 25m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/32f8ad91b2f04e4c9ef81b489c3b27c2 : SUCCESS in 56m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c5ae202e63ab9e9cf11c41b5b911e101a03374cc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-21 16:01:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6f7e7248c977be9778f5fa334c9b350e24471e4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 16:02:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d9c486d1d2b9532514707004b58b0c69d96db26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 21:48:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7ab75087ad2497bb8adaa50622892a9 : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ec0cd4c08b644a479bc52e47bc864b60 : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca7d85912281482bac68d9d43d956410 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d6ae53bb20645e9a30ae0a143afbf54 : SUCCESS in 7m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6319c1e268d44a55880f8baf6fe1fcf7 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77544805cc954b75a4a151f99f23d340 : SUCCESS in 6m 41s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b50accd3a330458b92d3539cb3a72220 : SUCCESS in 8m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2678a679fefa4544bbf8828320710fb6 : SUCCESS in 8m 01s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/355f8dda4b7346f3b36df67b7f4ac6ac : SUCCESS in 10m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0546e91546184b479baf569d63088826 : SUCCESS in 11m 27s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/124b67f4718b4dc7950f61f5a5b27f6b : SUCCESS in 5m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eafc18c4d9e845bbaa917460dc02dde8 : SUCCESS in 7m 39s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/25bb6742e90b462092f3fb70477055fd : SUCCESS in 12m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e96e52ad1af2468cbe96effc04ffd8c2 : SUCCESS in 12m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3aae513c4699433a84c8621b50ce111d : SUCCESS in 7m 37s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a440f99ac3954d9e9b6a193ebaaa7d7a : SUCCESS in 32m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fe12ede9b1194695b7f4c09e2d5a3bbf : SUCCESS in 14m 57s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a9197fdf147f47458e101a9aa1a2a055 : SUCCESS in 23m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/deefbe4676a24c249ed6d3df96994632 : SUCCESS in 24m 08s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/4cb44e26c49e44d0bbcea68fd302d4ac : SUCCESS in 13m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1799e877676b427fa7e1697c41426d21 : SUCCESS in 15m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d749e12b83b348c5aac421b4c437cb4b : SUCCESS in 17m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3c9324943782409c8a75e85f0e45dbf0 : SUCCESS in 19m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c160be9ad934a188d40f2dc01092223 : SUCCESS in 5m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2e421f1352db42c6baa77cf057a0a41d : SUCCESS in 11m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/32db133048574b29aebdff80d50646fc : SUCCESS in 10m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6abdc48dbaa84a498ac33b39a58f615b : SUCCESS in 1h 09m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6b5bee13c9cc470bb36847077a10f1f2 : SUCCESS in 1h 35m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/07061a2eea8447b9b12e8160d50255ab : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":1},{"id":"992a00d44801d0690662a64c7feaa8f227c0e08c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 21:48:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"86a3bc36061b6741a0df5734a4de589e1e05e0d2","revisions":{"86a3bc36061b6741a0df5734a4de589e1e05e0d2":{"kind":"REWORK","_number":1,"created":"2020-05-21 13:50:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/729964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/729964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/729964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/729964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/729964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/729964/1"}}},"commit":{"parents":[{"commit":"b2c5713f7ffa23a98465d224e5d8b596e3ea11a6","subject":"Merge \"Stop testing with pypy\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2c5713f7ffa23a98465d224e5d8b596e3ea11a6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-21 13:50:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-21 13:50:27.000000000","tz":0},"subject":"update constraint for sushy to new release 2.0.3","message":"update constraint for sushy to new release 2.0.3\n\nChange-Id: I8643c46c338560ff8293db504691ea8e38d4101b\nmeta:version: 2.0.3\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Commit: Mark Goddard \u003cmark@stackhpc.com\u003e\nmeta:release:Change-Id: I0c26817ffe901389c2abedf647fe8f3230307861\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+1: Julia Kreger \u003cjuliaashleykreger@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/86a3bc36061b6741a0df5734a4de589e1e05e0d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/86a3bc36061b6741a0df5734a4de589e1e05e0d2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
