)]}'
{"id":"openstack%2Frequirements~730098","triplet_id":"openstack%2Frequirements~master~I55acd9dd7d3ff8414b1621624d44091dea446d34","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I55acd9dd7d3ff8414b1621624d44091dea446d34","subject":"update constraint for openstack-release-test to new release 3.1.0","status":"MERGED","created":"2020-05-21 18:35:44.000000000","updated":"2020-05-22 02:12:57.000000000","submitted":"2020-05-22 02:11:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730098-1590113502891-da4d1bf7","meta_rev_id":"11bd200daef546d71993fd7e6f5df5e3dc98bbac","_number":730098,"virtual_id_number":730098,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-22 02:11:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-05-21 21:43:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-05-21 21:43:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 21:43:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-22 02:11:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"327c53c5686d50b55593d0d128fccb7882a6ce57","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-21 18:35:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1a697a21dad71fe1b2b9a207ce89c6c004b034c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 20:24:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/638b66fe39c44e008817f19dca1371e7 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/405e38bd2c834a86a08423ea3289c3e0 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03c0e883d68e4594b0d4431b742a938b : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/531712e0124d499897fd4ee5ed66b59b : SUCCESS in 5m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f3251f831ef04aaf965a7124ace72e13 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4530b1bf5fc54175b5929b1572c79a8c : SUCCESS in 5m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e2a6858b062242ccbf1cf700570f522a : SUCCESS in 10m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ab933f6ecfdc45cb8aa8f83eae375c45 : SUCCESS in 10m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0bd114c8b1e94931bb76599437414c13 : SUCCESS in 9m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/728f3148e2ab4eb395c05081611e2e8e : SUCCESS in 8m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9c487a2b74b0408e85cad78aad289a58 : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6df91f2302544862baacbd04e3682860 : SUCCESS in 8m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a03670f822304a55a39d9bb30ad109d3 : SUCCESS in 22m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bcf5067a80944ed2b389b795dfac64a7 : SUCCESS in 27m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d45e5411fbd1453db522102b62b035a9 : SUCCESS in 13m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/34cde72fbb7a4c64b617bceb53245961 : SUCCESS in 17m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/882db1e52faf496081eb04a17e68bac3 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/67df494b99f44a3bab16c82b143d2f2d : SUCCESS in 5m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3cf1a43bbf464653b2c9689c10a2313b : SUCCESS in 12m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf19aef34a3944a3833d9072c9ef8d3c : SUCCESS in 55m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2b65139f9174e46b164832e3992cf68 : SUCCESS in 1h 17m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d14ff761a9d4e0abe91d543a884cf1a : SUCCESS in 1h 41m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/e03b3e79bbb441d0bc35134bb228dcf5 : SUCCESS in 52m 26s","accounts_in_message":[],"_revision_number":1},{"id":"9bd827b9202cb8cd28bf876dc359d086c5e5d3d2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-21 21:43:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"83e18b4aed5773297a06da4a740c2d001a844eca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 21:43:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dccfa0e64f7b81d12e1a0d52d842ca19555879e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 02:11:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4db68a50d1564e1e822defcf1d832afc : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6ed4a6116e445709a1a3e27c4381500 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be838a52a681450d9f5e4ea290be8c93 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59ee85db2ace4217b182764b619c16cf : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/001b8abd23c445c69bbb8dd8d3eaf94c : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f5983d3fdd854744b980d63d567c8834 : SUCCESS in 3m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/31adc3feb2c5490eb696913b46133f1e : SUCCESS in 9m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/120ca1f2192a4858b47b2526cf08e5ba : SUCCESS in 9m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7f613fa75fba44529e18a614512ee5a2 : SUCCESS in 9m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3cd47e4dd2ba4cb598a67e40c0efb670 : SUCCESS in 5m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a367169ac5cb473688c84105d9fcaf4b : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3ee423effc3d40ed85352f13aa7cbbc4 : SUCCESS in 8m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3d7fd562a4c045ffbb914f26981812fc : SUCCESS in 13m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/914a80aab6ab468ebe1d853a1c71b5d5 : SUCCESS in 24m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ab20240cc138461bb9fbc8ee9fd7bfcd : SUCCESS in 14m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b8130d41e4e4ac2b370583bc6faf62b : SUCCESS in 19m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1764893470d641549bfd263da24cd618 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6be8a83bf05f4e3580911a7ab9f8480d : SUCCESS in 6m 11s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a0860a208b6a41b6af68d43a9b44d376 : SUCCESS in 11m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/035b1d1f4a8f4219a7d97cb9ce5efdb3 : SUCCESS in 10m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3cad854aa8d4f409c6900ad4e672599 : SUCCESS in 52m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/68f1d7b86818470593a2c07be8ccbb7d : SUCCESS in 1h 04m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24493fb9eb9745699bc1bad9fedf5be3 : SUCCESS in 1h 30m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/3ae9123027d849c9a55103288e5f657d : SUCCESS in 1h 04m 32s","accounts_in_message":[],"_revision_number":1},{"id":"e4eb46c131e2f6714581308f66058c3bfe500de1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 02:11:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4fd461c56f9e22c00a6f8550dbd005b93b604b48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 02:12:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d79d294bf521403d9d7bdd804cde5225 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3c6b3c34437b60aaaf02d2a3df486507d14d536","revisions":{"e3c6b3c34437b60aaaf02d2a3df486507d14d536":{"kind":"REWORK","_number":1,"created":"2020-05-21 18:35:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/730098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/730098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/730098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/730098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/730098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/730098/1"}}},"commit":{"parents":[{"commit":"8e5fda28894e8dc7860ec55a3526c96ff92206d1","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e5fda28894e8dc7860ec55a3526c96ff92206d1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-21 18:35:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-21 18:35:41.000000000","tz":0},"subject":"update constraint for openstack-release-test to new release 3.1.0","message":"update constraint for openstack-release-test to new release 3.1.0\n\nChange-Id: I55acd9dd7d3ff8414b1621624d44091dea446d34\nmeta:version: 3.1.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: yes\nmeta:release:Author: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta:release:Commit: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta:release:Change-Id: I17791ef7d56f0b4adae3e44a9c6eea86036594e1\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3c6b3c34437b60aaaf02d2a3df486507d14d536"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e3c6b3c34437b60aaaf02d2a3df486507d14d536"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
