)]}'
{"id":"openstack%2Frequirements~730148","triplet_id":"openstack%2Frequirements~master~Idbf8b76c2d53b61def7dc086e785e383c4e7e1db","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Idbf8b76c2d53b61def7dc086e785e383c4e7e1db","subject":"Cap few deps of keystoneauth for py3.5","status":"MERGED","created":"2020-05-21 23:50:10.000000000","updated":"2020-05-22 22:00:46.000000000","submitted":"2020-05-22 21:58:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730148-1590184733704-40f284a4","meta_rev_id":"2134981d29ffec62f6985f8c0e803a742fe872d4","_number":730148,"virtual_id_number":730148,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-22 21:58:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-05-22 20:06:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-05-22 19:37:10.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-05-22 20:06:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2020-05-22 19:28:42.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 19:37:10.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-05-22 20:06:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-22 21:58:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"603e53c91f3f524e84296b59f1f2804431384647","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-21 23:50:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f92863a315b786720fe6cb23f0b3b0e9a20889d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-21 23:52:08.000000000","message":"Patch Set 1: Workflow-1\n\nwe might need few more dep cap, testing the keystoneauth gate - https://review.opendev.org/#/c/730150/","accounts_in_message":[],"_revision_number":1},{"id":"d50cc5d48545aa24e85774fc50c9b5f7db06ad81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 01:04:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e6417117a014b0e95d9fe9c7e016c36 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23d1caf599c440cfad3c580d4c3469ff : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/488e401fefc34869a833cc6bf019d1ea : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71dd5f564b294dbf92ea004630e2204c : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19a54ea077a541e899e4d60c664d0b14 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/08128b172fe24cf6bd817d29b21bc92d : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b1eb8d9b70824d26b0e40f36634814d6 : SUCCESS in 8m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ecfaf25e389b4819ac7c67c33fb2a74c : SUCCESS in 14m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9764adfd03d249898f1f4e4cbf050c56 : SUCCESS in 9m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cd9eb30695f1444f991d2a4a97b759c4 : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e5f7691f2737413dad804445f90b9069 : SUCCESS in 9m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df86e2f3b5764600a19208869985b196 : SUCCESS in 7m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7792f69bea564537ad88f229ed8fc925 : SUCCESS in 22m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/431907ea459542cf9b7f8e220b7c20f9 : SUCCESS in 26m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f09fa40b8ca144e09c0cdbfd1f3d4e2b : SUCCESS in 12m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f4cfc43e3ffb422bb37e1ad588e5ceb6 : SUCCESS in 21m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1fde915937fe4dcb97048b8cb86976ab : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f23fba648e644b1385427b576399e0fa : SUCCESS in 6m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f719835bed2b48e8a2f9719b86a7a532 : SUCCESS in 11m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6c299d679d74d239db90f1291ac3981 : SUCCESS in 52m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/252d67c047294ebea0ac1f98c5976dca : SUCCESS in 1h 11m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/188f12be5ddf4e84bb220ed37bc0bf34 : SUCCESS in 1h 12m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/85406bb29bb14cb7a9025cee2150ab76 : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":1},{"id":"9db589ad66b431c6cd898e1bd14d9b59c501c04d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-22 01:13:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"77f4ae785c8cacc1afa2fc02aaaa618fe99ab111","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-22 01:13:33.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"1320fd164c5e68dcafafe410d349a34acc872e86","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-22 01:49:30.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"523661ceea5303eff1b55ddb3ab29243874c9fe1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-22 01:49:34.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"bbd5e7a6c6c8a8c65cad5d56e525eb37bf6d49e9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-22 02:36:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3be8907fd485f668ed8093804af7ea6fd5ffe6b5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-22 02:37:19.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"b6857920b596c8d19fa934779b147399303e8a1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 04:12:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e032b875db3d4251bb2f8b401a265f9a : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ba0b29ad0c4488b9245ca61e9be1c80 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad2691f3268e44feb940efefa6a4071a : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9d707f8ceb142afaad027b4e11754b6 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/89026c67182841e89591392c0f535ae9 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/220e5ab589bf436fb6692de5caa11a07 : SUCCESS in 4m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b135b0688a9a4f81bdb7d237a20286fb : SUCCESS in 8m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a6a551ff3a484a0db2ed501f6e287a81 : SUCCESS in 10m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/541a38caf66d4cbba2e47ba23d403f4d : SUCCESS in 8m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/69327ac6575c4a72a73cd260c72f491e : SUCCESS in 4m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2f12f2c6e68b4264b5d7ed895df429c8 : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fa39f880c0954b2d84392bb62a55b592 : SUCCESS in 7m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/42b4fb318ff649b9b8f782fb90468fdc : SUCCESS in 22m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ff23af1442074dbbba79fc443ea164ae : SUCCESS in 24m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4ca05671780d414fa6e1999d18d5688f : SUCCESS in 17m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/20263052e29f4551a214cf0fe9d2f61e : SUCCESS in 15m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/93a975ccae254c8086054724423b1f8d : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/52c9947286f14eaab1b076a880346cb1 : SUCCESS in 7m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/339118a75bc84cb8b93cfac3db86a195 : SUCCESS in 10m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cbacf7e6ff54fb1b344d22622d033be : SUCCESS in 48m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa019b19a8004965abc1775853aa78a6 : SUCCESS in 1h 06m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5040a7d4a9344516a57217a145358cfc : SUCCESS in 1h 30m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/18bd6f0a5bca4b0a9dda558b50406993 : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":4},{"id":"a1b62b115615386f26ea77d7e8fb452cb35fa0f8","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-05-22 14:39:47.000000000","message":"Patch Set 4: Code-Review+2\n\nPut a depends-on in https://review.opendev.org/728920","accounts_in_message":[],"_revision_number":4},{"id":"328726fc59a4667e1a01d85f6add2ef59b8a6101","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-22 16:27:56.000000000","message":"Patch Set 4: -Workflow\n\nthis made py35 green - https://review.opendev.org/#/c/730150/ this is good to go.\n\nBut requirement-checks job failing for mismatch line for python version cap, do we need same update manually on ksa side also or wait for bot patch.\n- https://review.opendev.org/#/c/727498/5","accounts_in_message":[],"_revision_number":4},{"id":"5dd6019ea9055af430b99c80c6dc0306f711b114","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-22 18:55:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"43a9196bcc6067367460df066a6c72c24eb65825","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-22 19:03:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f7771b650210d92ba992f8139dc741443a6ffab3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-22 19:05:35.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"650ed0caab4768b21c58cb02b865fdf27aca7e03","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-22 19:06:59.000000000","message":"Patch Set 5: Workflow-1\n\nchecking req check job here if we need to soften req check for py3.5 - https://review.opendev.org/#/c/727498/5","accounts_in_message":[],"_revision_number":5},{"id":"08e1ed420cf9c4e9398818a14c9505d87822021d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-05-22 19:28:42.000000000","message":"Patch Set 5: -Workflow\n\nok, all good on req check jobs also now.","accounts_in_message":[],"_revision_number":5},{"id":"e81a34f025edba2dccfa92948b29a6b4ad547b4f","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-05-22 19:37:10.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0ad619934cdfa3419afac9a675bf4e3a7ad8056e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-22 20:06:31.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nI think this will be good","accounts_in_message":[],"_revision_number":5},{"id":"121b50a3042d30188422045af20205bf64a7d42e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 20:21:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0b6c69061e44eceb0266443fc29e264 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1b6b8668a084fbba51fa2494cf33576 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c0438a6c58f4b008673e5cb9d1febcc : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ea33a11c1a1487d9915aa8e94c30935 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f766b425501431dacedbe97fc786713 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f80cadfda005487397bcf3768c05b107 : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1fd43c6a447042fc99937f5f7e275e19 : SUCCESS in 8m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d8de278d7dc0403fb8d01b9dca042920 : SUCCESS in 9m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/609aa9de74e0443e9275f3624bff804e : SUCCESS in 11m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3829235b041d460a83e2967bb7807263 : SUCCESS in 5m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c6e86d5e568e43d29de190e24037b966 : SUCCESS in 11m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/07e3a0fba23d4d878b7bcfe72983249a : SUCCESS in 7m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cff17ef1eebd4d1a80447c7cc9c9eda9 : SUCCESS in 16m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/73e2024e6d284633b6bdb6116682288e : SUCCESS in 33m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bc046f61b2714c4e8cd041b8e00e1ee8 : SUCCESS in 16m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e475defc3b5d4b4ab9fd497173250c04 : SUCCESS in 17m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ead75e62592d44239169b71b298b3ba6 : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/001a1b381f0240c59a2196b22be3a908 : SUCCESS in 6m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fe9dcc6318504e66a6e3c3f2166774c5 : SUCCESS in 15m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/705af0372cab457ebce068e184627ac1 : SUCCESS in 1h 00m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4c7e125c6b6040cbb8208dad6684740c : SUCCESS in 1h 05m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59621c49a3fd42149ecc85e52be65629 : SUCCESS in 1h 14m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/f9809214664544099bd9cf409cebf62b : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":5},{"id":"e3af5d61a8bb1a109875870a270f1ddcc63fcd6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 20:21:31.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bfc3d6f354f66fad3a6ef4e796b526e5205f4c74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 21:58:53.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8f46e5d399d4a3c8029b395843b0870 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40b48466f3c04eb48af4a95191a5a658 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/317e032ce72a414cb31f78f3e3ab2fef : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00104da601b642bd8be994ce06b5e0e0 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f927de277c9b49f4bc595ddbcc4d01d5 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fecdc40fbb6b4563bab742aa6654bcf2 : SUCCESS in 4m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d459a8abf9724e3a906e4e357015238a : SUCCESS in 13m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/608f6e4fb04f4af6a41db45ed1f46e59 : SUCCESS in 9m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2eea35a8ea3f43aeb83eb4468c0cc472 : SUCCESS in 13m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6c500861dc894665b8e340e009d7ecc1 : SUCCESS in 6m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6c5211fcb6c3468a824ff317ddffd59e : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4d9eabbc4e674ed9b9bfcf16bf5d2b85 : SUCCESS in 13m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/daa8bd4203a34dc3b771eae39b3d68fa : SUCCESS in 14m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/83463c60ca1e43189ba3b839f837e358 : SUCCESS in 29m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ec24fc529af041a2bcec4dcdf5eec1f8 : SUCCESS in 17m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/362f370c01ef4406a5cfc84ea613c67b : SUCCESS in 15m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4193e051548b44fc84ba3f28289aae83 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f7e1aade43444218a5954d115c9985d : SUCCESS in 7m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e17bca16ba0b4e2a947e3250b520f5bc : SUCCESS in 14m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/14b08d72f2c54159baf6e1beb264f26f : SUCCESS in 9m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42e61280f79b4181a18e545f9ce49cbe : SUCCESS in 56m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dcaecfbc27274bdcbbc188f3e7a8ed15 : SUCCESS in 1h 14m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d20f64e653514dbe8aac1772ef4aa4ce : SUCCESS in 1h 31m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/c25bfd731aff493ab0514928d153e522 : SUCCESS in 1h 05m 29s","accounts_in_message":[],"_revision_number":5},{"id":"5b36d044e664e6a81842ff4535bb7fab041094f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 21:58:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"2134981d29ffec62f6985f8c0e803a742fe872d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 22:00:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b9f815636a41dd86e614af74b9cad4 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4602eee943f0dba286477784ec7d362c6c2e3146","revisions":{"37a7739bde4f8b5fbb50a911af4a2eb18f59ec05":{"kind":"REWORK","_number":1,"created":"2020-05-21 23:50:10.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/48/730148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/730148/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/730148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/730148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/730148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/730148/1"}}},"commit":{"parents":[{"commit":"118fbacc6037c5c88d90743e88cca81dc86af007","subject":"Merge \"Cap zipp for py3.5\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/118fbacc6037c5c88d90743e88cca81dc86af007"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-21 23:46:52.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-21 23:46:52.000000000","tz":-300},"subject":"Cap oslotest for py3.5","message":"Cap oslotest for py3.5\n\nFew branchless projects like Keystoneauth, OpenStackSDK etc\nkeeping the support of py3.5 and have voting py3.5 job.\n\nTo fix the currently failing py3.5 job on keystonauth,\nwe need to have oslotest cap for \u003c py3.6. We need to keep it\nuntil we completely drop py3.5 from OpenStack.\n- https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html#projects-keeping-py3-5-support\n\nChange-Id: Idbf8b76c2d53b61def7dc086e785e383c4e7e1db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/37a7739bde4f8b5fbb50a911af4a2eb18f59ec05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/37a7739bde4f8b5fbb50a911af4a2eb18f59ec05"}]},"branch":"refs/heads/master"},"3e36c9f6b3ea864e2f05ada6e8280db3be0e64af":{"kind":"REWORK","_number":2,"created":"2020-05-22 01:13:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/48/730148/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/730148/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/730148/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/730148/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/730148/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/730148/2"}}},"commit":{"parents":[{"commit":"118fbacc6037c5c88d90743e88cca81dc86af007","subject":"Merge \"Cap zipp for py3.5\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/118fbacc6037c5c88d90743e88cca81dc86af007"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-21 23:46:52.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-22 01:13:18.000000000","tz":0},"subject":"Cap few deps of keystoneauth for py3.5","message":"Cap few deps of keystoneauth for py3.5\n\nFew branchless projects like Keystoneauth, OpenStackSDK etc\nkeeping the support of py3.5 and have voting py3.5 job.\n\nTo fix the currently failing py3.5 job on keystonauth,\nwe need to have kesutoneauth deps cap for \u003c py3.6. We need to keep it\nuntil we completely drop py3.5 from OpenStack.\n- https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html#projects-keeping-py3-5-support\n\nChange-Id: Idbf8b76c2d53b61def7dc086e785e383c4e7e1db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e36c9f6b3ea864e2f05ada6e8280db3be0e64af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e36c9f6b3ea864e2f05ada6e8280db3be0e64af"}]},"branch":"refs/heads/master"},"da7951e9c851ca6adb0f32fc699b1eabbce97cd5":{"kind":"REWORK","_number":3,"created":"2020-05-22 01:49:30.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/48/730148/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/730148/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/730148/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/730148/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/730148/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/730148/3"}}},"commit":{"parents":[{"commit":"118fbacc6037c5c88d90743e88cca81dc86af007","subject":"Merge \"Cap zipp for py3.5\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/118fbacc6037c5c88d90743e88cca81dc86af007"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-21 23:46:52.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-22 01:49:22.000000000","tz":0},"subject":"Cap few deps of keystoneauth for py3.5","message":"Cap few deps of keystoneauth for py3.5\n\nFew branchless projects like Keystoneauth, OpenStackSDK etc\nkeeping the support of py3.5 and have voting py3.5 job.\n\nTo fix the currently failing py3.5 job on keystonauth,\nwe need to have kesutoneauth deps cap for \u003c py3.6. We need to keep it\nuntil we completely drop py3.5 from OpenStack.\n- https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html#projects-keeping-py3-5-support\n\nChange-Id: Idbf8b76c2d53b61def7dc086e785e383c4e7e1db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da7951e9c851ca6adb0f32fc699b1eabbce97cd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da7951e9c851ca6adb0f32fc699b1eabbce97cd5"}]},"branch":"refs/heads/master"},"512b53e9ec626a1dbac1d76580eb7bbbffc859ff":{"kind":"REWORK","_number":4,"created":"2020-05-22 02:36:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/48/730148/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/730148/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/730148/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/730148/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/730148/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/730148/4"}}},"commit":{"parents":[{"commit":"118fbacc6037c5c88d90743e88cca81dc86af007","subject":"Merge \"Cap zipp for py3.5\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/118fbacc6037c5c88d90743e88cca81dc86af007"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-21 23:46:52.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-22 02:34:53.000000000","tz":-300},"subject":"Cap few deps of keystoneauth for py3.5","message":"Cap few deps of keystoneauth for py3.5\n\nFew branchless projects like Keystoneauth, OpenStackSDK etc\nkeeping the support of py3.5 and have voting py3.5 job.\n\nTo fix the currently failing py3.5 job on keystonauth,\nwe need to have kesutoneauth deps cap for \u003c py3.6. We need to keep it\nuntil we completely drop py3.5 from OpenStack.\n- https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html#projects-keeping-py3-5-support\n\nChange-Id: Idbf8b76c2d53b61def7dc086e785e383c4e7e1db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/512b53e9ec626a1dbac1d76580eb7bbbffc859ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/512b53e9ec626a1dbac1d76580eb7bbbffc859ff"}]},"branch":"refs/heads/master"},"4602eee943f0dba286477784ec7d362c6c2e3146":{"kind":"REWORK","_number":5,"created":"2020-05-22 19:05:35.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/48/730148/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/730148/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/730148/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/730148/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/730148/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/730148/5"}}},"commit":{"parents":[{"commit":"118fbacc6037c5c88d90743e88cca81dc86af007","subject":"Merge \"Cap zipp for py3.5\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/118fbacc6037c5c88d90743e88cca81dc86af007"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-21 23:46:52.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-05-22 19:05:32.000000000","tz":0},"subject":"Cap few deps of keystoneauth for py3.5","message":"Cap few deps of keystoneauth for py3.5\n\nFew branchless projects like Keystoneauth, OpenStackSDK etc\nkeeping the support of py3.5 and have voting py3.5 job.\n\nTo fix the currently failing py3.5 job on keystonauth,\nwe need to have kesutoneauth deps cap for \u003c py3.6. We need to keep it\nuntil we completely drop py3.5 from OpenStack.\n- https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html#projects-keeping-py3-5-support\n\nChange-Id: Idbf8b76c2d53b61def7dc086e785e383c4e7e1db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4602eee943f0dba286477784ec7d362c6c2e3146"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4602eee943f0dba286477784ec7d362c6c2e3146"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
