)]}'
{"id":"openstack%2Frequirements~730304","triplet_id":"openstack%2Frequirements~master~I5e0e8e6a5af0f439a0b3e972d9a9ed07fcaaddcd","project":"openstack/requirements","branch":"master","topic":"xdist","hashtags":[],"change_id":"I5e0e8e6a5af0f439a0b3e972d9a9ed07fcaaddcd","subject":"Add pytest-xdist to global requirements","status":"MERGED","created":"2020-05-22 11:29:00.000000000","updated":"2020-05-28 00:31:34.000000000","submitted":"2020-05-28 00:27:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730304-1590625644512-649d007f","meta_rev_id":"83e4589d9cb8ea673ee8824dc9391dc42f0dd6fa","_number":730304,"virtual_id_number":730304,"owner":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 00:27:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-05-26 15:59:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-05-27 21:12:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-27 12:18:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2020-05-27 02:23:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-05-27 21:12:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 15:59:35.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-27 02:23:55.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-05-27 12:18:46.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2020-05-27 21:12:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-28 00:27:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac34bcb266ed8feeb280bba15e35fd0ae02179df","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-05-22 11:29:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aed7159d0dbf16a04bdd0aa7efdb99122d802b5b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-22 11:38:51.000000000","message":"Patch Set 1: Code-Review-1\n\nupper-constraints.txt needs to be updated too.","accounts_in_message":[],"_revision_number":1},{"id":"e66a1548ff3a13c39ba1fe31ad7b3321e6c7a89e","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-05-22 12:09:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3d32521584de561eb5fd11a8e457d04a8a64246f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 13:35:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9831a74814a4176b8144d58acf9b360 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e9283fa37d145cb86d911955a7139e4 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbd19a38d7ac47548380a974a5d5d68e : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab0ad9ed369e4e7a8939aa3e45162431 : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b910130708a44f53bcb9c98f3f44fdec : FAILURE in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1916c31c64ba40cbaceca185076e04f3 : SUCCESS in 3m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/985e43121a6e49b4818cf9ea30d30f9d : SUCCESS in 8m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/73248ceb269d4b6bafb4d77057c3e0dd : SUCCESS in 12m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6d2c3def3b1341cfa0d083ac80cf478f : SUCCESS in 9m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0e64b63bed97412a85c2f1ae4508eeec : SUCCESS in 4m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/39006d7c1af54f38bc8e515be487eeae : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a2110ea1ef4b4df3bece95d19a23d1ff : SUCCESS in 7m 16s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/173b71aae3a04f6287d0af3bf4ade124 : SUCCESS in 32m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/15af4a692cf449e990fa3435f3b6ad2f : SUCCESS in 29m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ee127e1d8d874cbaa0c8fa63f191d58c : SUCCESS in 14m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9039ae038044442e946b6dcacba70b31 : SUCCESS in 21m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/446c819fb3a64358a8e66aaae475e7e8 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ecd4e52cbf9c4173a778dbc6a1f5e561 : SUCCESS in 6m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/37db490c20b241a19a032feb55d32cb7 : SUCCESS in 14m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f18fd41c3774c9fa0de5a8891c0cd10 : SUCCESS in 51m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5fae65aac5dd4e6aa97fc965a386046c : SUCCESS in 1h 12m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c4e35740b274980afdbf0c5204efd17 : SUCCESS in 1h 24m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/b1c69a7005ae4d00a18d8931fd7348df : SUCCESS in 1h 08m 13s","accounts_in_message":[],"_revision_number":2},{"id":"1506875422e0e2d1bd5c07235fb556ab70b1dd1e","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-05-22 13:45:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"176f7f1a83d483526ebc81aa0360144935b1ead7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 15:17:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eba0d16789f64e19a384883cb5acc279 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ab503c093d4501a4eec08917030ab1 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c91ecc8e44b41028f2d5f16760f2692 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c6df3ff56584db88e999e4f98300452 : SUCCESS in 4m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82ae515b0f9d465eab8b891d79ab6948 : FAILURE in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b35989ca67d740d59e9c4d138f720ef2 : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e6b98b08077f4479a5549441991723e2 : SUCCESS in 8m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/de706ae7234748aba22649c835876b39 : SUCCESS in 11m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d64070a6b25d4a28a1b378554879b87b : SUCCESS in 11m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/380a6a38967a498288ea3727cf7e5896 : SUCCESS in 6m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4ae13d70813e4740ba19616afe86ab4c : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f41bcef555564c9fa7a664cce3e261f0 : SUCCESS in 9m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c3257fad60a24aba9a7e88a6725f6a38 : SUCCESS in 17m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b703921ed44140c9a7aa933076516f8d : SUCCESS in 26m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b2b63ea20f494e299b01aaadf6e7eb43 : SUCCESS in 13m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f2030af07a64c95a1994153deb21a33 : SUCCESS in 20m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d16932076885478aab718bdc3a7bccb3 : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/29c6ea82393e493e838f062e0c78872f : SUCCESS in 8m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cb1f6ee317804ce2949a0bc6b949b02d : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0bbf5ce9994a47d89e38d21e06b70da4 : SUCCESS in 53m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fab80e679eee4c82b231b7460630cdce : SUCCESS in 1h 18m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf5e09e2382c4abab8eedbcecb3cd6fe : SUCCESS in 1h 28m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/9253ec4dc3074ba58bc789434436a96e : SUCCESS in 1h 14m 49s","accounts_in_message":[],"_revision_number":3},{"id":"a74186a053aefcfbc1046034ef2a7a442eeda166","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-05-22 15:41:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c4737ebe2d8f65ebcb862c9026ef7aa0f281fc29","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-22 15:48:24.000000000","message":"Patch Set 4:\n\nI think we still need the questions answered in the commit message.\n\nhttps://docs.openstack.org/project-team-guide/dependency-management.html#for-new-requirements","accounts_in_message":[],"_revision_number":4},{"id":"929b2bd285ba4c13e08674446861e5dbde9ddd55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 17:01:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7489e90b3e1440048af128718b9b7406 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80d4245e10e34272ada1eaa9b8bffb85 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfe978b7cf7f468d945b50f40530c284 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bf009d2d6494bd38efca3a97f75e2c3 : SUCCESS in 8m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0416b90276c54f9aaf73442f156b6ced : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/03fae16ada64439aa4b5380d2c4be771 : SUCCESS in 5m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1730e24a7c4c4ebdbb80104952d72454 : SUCCESS in 12m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7d53c32d1750403289b0a49947402459 : SUCCESS in 13m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d1e47a444a1045bc91d2428e32cd5dcb : SUCCESS in 12m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/75f10f4eecae4024819a1aa8453f1a33 : SUCCESS in 7m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/585f4c06bec0474d955519867f3eac87 : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eaef5f8633f4496eb417fb7770e6f1b8 : SUCCESS in 11m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7ea217c46bb74b17aee6bf353a92f9d7 : SUCCESS in 22m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0d14ab1d188444cd8fc2bd62f9b1d599 : SUCCESS in 28m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d603c795250944e39bc52e89a92f1878 : SUCCESS in 17m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2d06d6f5635144f393bc71c34d14334b : SUCCESS in 21m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f58f0757f2f9476da0872b4873d034e0 : SUCCESS in 7m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d52d6537f394242972441fe0a734e7d : SUCCESS in 7m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e40f2038ba8047e1b6c6c30550f1cd78 : SUCCESS in 13m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae53ab68542f4b8094e7101bf89b000e : SUCCESS in 54m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3a1c1e2fc6804334852df42482c6bfcb : SUCCESS in 1h 04m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ddf7722e56b4d34b855d3d08f15dd34 : SUCCESS in 1h 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/f28a6f54b9be466eaa3b0f842cb40356 : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":4},{"id":"535d0713131f82320eb9931ac2712039e407dc8c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-23 06:03:56.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"4dcfae5ed39787e4c7e335aff8f8fbcd39e8fc04","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-05-25 12:27:01.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"48ce42942ee54d5318cf72254f93393f3a31c70d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 13:50:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d2de7b2047042c0a0c45d335290ebe0 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d069747dadb4f4c948bd0e75de16e11 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05aae699d52c413a913a447246583504 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58873251af2d4ede9ebdfd5cf7a09c23 : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dd4f72738230477da71ed8203fbf45d2 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6eec9b0954e04f458261e05bf2f875a7 : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/53dc49fdc227425bb983ff8c555a50ec : SUCCESS in 8m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/361d7f761ddf40a68af33f35ac779182 : SUCCESS in 8m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/76b4b5ccb6574fd0a9750348c22234f7 : SUCCESS in 8m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4603d03e945140fdb456a68dcc31a80c : SUCCESS in 5m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/32799043568b4b84a148f37de4e6d460 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f7760323be364001a52fe24271e1fda5 : SUCCESS in 7m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/34f706160d994807b32160e5df37b87f : SUCCESS in 12m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/42c72bf427a84b8380fd04883c862634 : SUCCESS in 31m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/79b184c7c5ca453587c06517628a5efe : SUCCESS in 13m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d9be893ef4341a98f2fa0039ebf968a : SUCCESS in 21m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/985de649356e487ea181139469b1bfda : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c8033122f27e4962a009caa652f563b7 : SUCCESS in 6m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6ed327734b95463da5d22851fef034b1 : SUCCESS in 10m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb0f99e46b23454c99a64095f7c0fb66 : SUCCESS in 53m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/46da80ed957d40f78674908289d684f3 : SUCCESS in 1h 08m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04f06766c3784022b067709bdc46ac13 : SUCCESS in 1h 18m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/659609222e7a477ba9e0b0b769746977 : SUCCESS in 1h 04m 37s","accounts_in_message":[],"_revision_number":5},{"id":"badcf7267d2b1723951efbcaeb7851d7119b12ad","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-26 15:59:35.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7ac5c99fd88565e71d0f68413e343e4be26df31f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-05-27 02:23:55.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"0a5f615c36201b6ad8d74326ac46c0373160fc5f","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-05-27 12:18:46.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"2b3ad3253436c9e79dd0cd27204795fe39cf3745","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-27 21:12:42.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"0420cdde1a6089815013a89aa301b87408de51b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 21:13:10.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fcbdb16bdf87d2a51bbccdf8196fab1fda917535","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 00:27:24.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38cf6481a9e740da881b34b661d881ca : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27552384191f40a1a8d9f54b5e640c1b : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89a3832c6b694eb2bc435b6dfa83de29 : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/094a2dd45d484e6e9f4d02f47e5d272f : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c99d84b49b04f08b98be1f9c8d37851 : SUCCESS in 6m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4b7259533c854d81af182b82470603b5 : SUCCESS in 6m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/446e1054d6f740eab5dd5ec2845de287 : SUCCESS in 9m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0973c21cf4c1408f93bdbe0413f904ff : SUCCESS in 11m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/92b8514980314e54829153cc76d31941 : SUCCESS in 12m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/39ecc7ff3e88435abe343bc1a0fe37a5 : SUCCESS in 7m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/951f7258a7d3411ea21a956a1e5faa7d : SUCCESS in 13m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/97f65c0b42094598af66370a190bd61f : SUCCESS in 13m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8e3ce3852b8e4918afc3e6bb287bf096 : SUCCESS in 22m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9282bd997e804ac589214e3e3a935ae7 : SUCCESS in 25m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1d64d67c2af3470e9e901d821e018fd5 : SUCCESS in 20m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/044c694d03e543db91964a3dd7a246c2 : SUCCESS in 22m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/18f1c3f7546a4b158ca454426c74a7b4 : SUCCESS in 7m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f38e9cb407441c683dd92e0b2a9e14a : SUCCESS in 9m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/1f2dcd66d50a4436a7c777f9a2e4fae2 : SUCCESS in 13m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d8e8ff93ba874b9282bb0a9b333bd8a5 : SUCCESS in 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2471bd652bf4035b8dbb8acfa546b72 : SUCCESS in 1h 05m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e161f2388654c36babb2a1b1407f6b1 : SUCCESS in 1h 08m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/374f61d7e4f8413f96b572f9d33328ae : SUCCESS in 1h 27m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/35178db4f84845939b8bebe019113e46 : SUCCESS in 1h 11m 38s","accounts_in_message":[],"_revision_number":5},{"id":"ed0bc921f5f63d17fcf5286c956f82854453037b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 00:27:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"25df7e51afbf125f852ff581440a0ba81eb4cb9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 00:31:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ea95b77015e4197ade7bd41313b60e1 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9b1ba719b75e1856ff6d897df847e401225babe2","revisions":{"36b62446b479a9664c7475eff4211622c6e6cbd0":{"kind":"REWORK","_number":1,"created":"2020-05-22 11:29:00.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/04/730304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/730304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/730304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/730304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/730304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/730304/1"}}},"commit":{"parents":[{"commit":"3b26fe47a9f897c46c2b58d069faf19ea24f09f7","subject":"Merge \"update constraint for openstack-release-test to new release 3.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b26fe47a9f897c46c2b58d069faf19ea24f09f7"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-05-22 11:26:39.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-05-22 11:28:58.000000000","tz":180},"subject":"Add pytest-xdist to global requirements","message":"Add pytest-xdist to global requirements\n\npytest was introduced in [1] as horizon requirements. Horizon\nwants to speedup tests execution time now, so we need this\nplugin to be included into requirements.\n\n[1] Idba192945ee8980b6ee2d49ce9651dbedb479609\n\nChange-Id: I5e0e8e6a5af0f439a0b3e972d9a9ed07fcaaddcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36b62446b479a9664c7475eff4211622c6e6cbd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36b62446b479a9664c7475eff4211622c6e6cbd0"}]},"branch":"refs/heads/master"},"eb9a6df7fe29c5496635f970b3c1ea1292040eaf":{"kind":"REWORK","_number":2,"created":"2020-05-22 12:09:06.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/04/730304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/730304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/730304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/730304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/730304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/730304/2"}}},"commit":{"parents":[{"commit":"3b26fe47a9f897c46c2b58d069faf19ea24f09f7","subject":"Merge \"update constraint for openstack-release-test to new release 3.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b26fe47a9f897c46c2b58d069faf19ea24f09f7"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-05-22 11:26:39.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-05-22 12:08:15.000000000","tz":180},"subject":"Add pytest-xdist to global requirements","message":"Add pytest-xdist to global requirements\n\npytest was introduced in [1] as horizon requirements. Horizon\nwants to speedup tests execution time now, so we need this\nplugin to be included into requirements.\n\n[1] Idba192945ee8980b6ee2d49ce9651dbedb479609\n\nChange-Id: I5e0e8e6a5af0f439a0b3e972d9a9ed07fcaaddcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb9a6df7fe29c5496635f970b3c1ea1292040eaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb9a6df7fe29c5496635f970b3c1ea1292040eaf"}]},"branch":"refs/heads/master"},"d5a03131eb6b4fd87261a2ec36bb4623951dceec":{"kind":"REWORK","_number":3,"created":"2020-05-22 13:45:29.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/04/730304/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/730304/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/730304/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/730304/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/730304/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/730304/3"}}},"commit":{"parents":[{"commit":"3b26fe47a9f897c46c2b58d069faf19ea24f09f7","subject":"Merge \"update constraint for openstack-release-test to new release 3.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b26fe47a9f897c46c2b58d069faf19ea24f09f7"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-05-22 11:26:39.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-05-22 13:45:18.000000000","tz":180},"subject":"Add pytest-xdist to global requirements","message":"Add pytest-xdist to global requirements\n\npytest was introduced in [1] as horizon requirements. Horizon\nwants to speedup tests execution time now, so we need this\nplugin to be included into requirements.\n\n[1] Idba192945ee8980b6ee2d49ce9651dbedb479609\n\nChange-Id: I5e0e8e6a5af0f439a0b3e972d9a9ed07fcaaddcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5a03131eb6b4fd87261a2ec36bb4623951dceec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d5a03131eb6b4fd87261a2ec36bb4623951dceec"}]},"branch":"refs/heads/master"},"6e5fe95d85551274ff569a3c52e18f22c161093c":{"kind":"REWORK","_number":4,"created":"2020-05-22 15:41:50.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/04/730304/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/730304/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/730304/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/730304/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/730304/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/730304/4"}}},"commit":{"parents":[{"commit":"3b26fe47a9f897c46c2b58d069faf19ea24f09f7","subject":"Merge \"update constraint for openstack-release-test to new release 3.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b26fe47a9f897c46c2b58d069faf19ea24f09f7"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-05-22 11:26:39.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-05-22 15:41:40.000000000","tz":180},"subject":"Add pytest-xdist to global requirements","message":"Add pytest-xdist to global requirements\n\npytest was introduced in [1] as horizon requirements. Horizon\nwants to speedup tests execution time now, so we need this\nplugin to be included into requirements.\n\n[1] Idba192945ee8980b6ee2d49ce9651dbedb479609\n\nChange-Id: I5e0e8e6a5af0f439a0b3e972d9a9ed07fcaaddcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6e5fe95d85551274ff569a3c52e18f22c161093c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6e5fe95d85551274ff569a3c52e18f22c161093c"}]},"branch":"refs/heads/master"},"9b1ba719b75e1856ff6d897df847e401225babe2":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-05-25 12:27:01.000000000","uploader":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"ref":"refs/changes/04/730304/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/04/730304/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/730304/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/730304/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/04/730304/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/04/730304/5"}}},"commit":{"parents":[{"commit":"3b26fe47a9f897c46c2b58d069faf19ea24f09f7","subject":"Merge \"update constraint for openstack-release-test to new release 3.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3b26fe47a9f897c46c2b58d069faf19ea24f09f7"}]}],"author":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-05-22 11:26:39.000000000","tz":180},"committer":{"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","date":"2020-05-25 12:18:40.000000000","tz":180},"subject":"Add pytest-xdist to global requirements","message":"Add pytest-xdist to global requirements\n\npytest was introduced in [1] as horizon requirements. Horizon\nwants to speedup tests execution time now, so we need this\nplugin to be included into requirements.\n\nIs the library actively maintained?\nAccording to GitHub [2] project has active community with\nregular releases.\n\nIs the library good code?\nIt has a good number of unit tests and uses black for code\nformatting.\n\nLicense: MIT\n\nIs the library already packaged in the distros we target?\nYes, there are packages in Ubuntu and Fedora repositories.\n\nIs the function of this library already covered by other\nlibraries in global-requirements.txt?\nNo.\n\nIs the library required for OpenStack project or related\ndev or infrastructure setup?\nIt\u0027s used only for tests.\n\n[1] Idba192945ee8980b6ee2d49ce9651dbedb479609\n[2] https://github.com/pytest-dev/pytest-xdist\n\nChange-Id: I5e0e8e6a5af0f439a0b3e972d9a9ed07fcaaddcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b1ba719b75e1856ff6d897df847e401225babe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b1ba719b75e1856ff6d897df847e401225babe2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
