)]}'
{"id":"openstack%2Frequirements~730850","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I16a4438d059014e70273c59de0909a8d24155f88","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I16a4438d059014e70273c59de0909a8d24155f88","subject":"update constraint for tripleo-common to new release 12.4.0","status":"MERGED","created":"2020-05-26 15:20:52.000000000","updated":"2020-05-26 23:23:10.000000000","submitted":"2020-05-26 23:23:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730850-1590535390899-48d602ed","meta_rev_id":"e29f9c02a23a42a2c6ae72a2a6232a879e3af955","_number":730850,"virtual_id_number":730850,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 23:23:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-26 15:31:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-26 20:58:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-26 20:58:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 15:31:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-26 20:58:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-26 23:23:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0054ab5b7a835a1db5435bace743a26cf8e65f53","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-26 15:20:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35ff2e41e15cdc957ac4480b9b3f6119c15c6d43","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-26 15:31:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"42fce5ec97567db9bf38bb4efaeeb96bbfa46ca8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:47:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc2575edd74d4573b79e37d348729db1 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb02adb3316f419eaa66f6fe3fc7b39a : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/159e14d927924126b243bb6a893d3907 : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b344c33881b46f2b6a7e0f1f467bbd6 : SUCCESS in 5m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1c0816037a27464fba133aee98e4af2b : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e42b69f855dd4b5484203c18c24f6ff9 : SUCCESS in 10m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3f1b58a4f0ec4ab49c291e10d267ba1f : SUCCESS in 10m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4e3308047ccc4cdf980a3a797f763573 : SUCCESS in 6m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4cb3ad11234945ae9684096d02cccc70 : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/de3261ff03e84d739aa01edc62ff476b : SUCCESS in 9m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ea87ccd8265046a59e83b6f546fc128d : SUCCESS in 14m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/02c1003b2fd8442f9fd8171cdfef5f97 : SUCCESS in 33m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5eab35b28e88417898165ec56a242adf : SUCCESS in 14m 43s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/04e395f5189c4f78b3a1410e283ffbdf : SUCCESS in 20m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a646a1f9b5bf4194bdb4c4787198b02c : SUCCESS in 22m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/048b493ab2bf4610b658879d9f2c8234 : SUCCESS in 5m 24s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9bb39bfff066424b995bd6b3660d02da : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ccb44f8ea9bd43ed94cc81ecd4c24ba0 : SUCCESS in 9m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/16239ab3d8dd4c229a5049b99417b952 : SUCCESS in 10m 24s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f40b360d97f1463f97b5b7372be8c66e : SUCCESS in 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/312b14858472451eb44a0f3540811c40 : SUCCESS in 53m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca534b04703046c6b5b988da66a8db5b : FAILURE in 10m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/668dd999586d4d92b377f3516335912a : SUCCESS in 1h 36m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/2a0109b8cfc64297b30199d565713f98 : SUCCESS in 1h 06m 52s","accounts_in_message":[],"_revision_number":1},{"id":"eacc294b861fd06b508d0ca1681cdcafe787e077","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-26 17:58:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"067301a5bd3dd790827207a58e882639536ce2d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 19:26:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1cfd1cdd2c14b24b7ca4cca22ab560c : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4258694eaea41e5b6cb2bca1a60bdf4 : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/96379a5325b042dd991e096f8a17b365 : SUCCESS in 4m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45f36b5086b14b4f9975632ec339c67c : SUCCESS in 6m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6e1a1cdea63c45e3bdd83354b4183f47 : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/296ae8bae2a94b4a8788654cc39f68a2 : SUCCESS in 9m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b6733010c4df437fb79ce0a2c349b2cc : SUCCESS in 11m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4eee53fd2e404cf889ebd2719b660364 : SUCCESS in 6m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0ad93beb19c842499e957e18fed25487 : SUCCESS in 13m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d1a495d20f69472aa6b103c157555cca : SUCCESS in 9m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/280a3b1f80d14278bb5336cd1478baf7 : SUCCESS in 26m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d3748c44cfd640d8813ea27312f77cda : SUCCESS in 29m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/67e186fed6a74f1f9c283bfb3b707154 : SUCCESS in 16m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/137d2a840a4842e98c48e778d4ae319f : SUCCESS in 18m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/68c8e1c6723642f5addd751f4fd2101d : SUCCESS in 31m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/74e0bc41621745e8ad08cd833b12799d : SUCCESS in 6m 22s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/210b6b4317064270ad7ce1c7ee8669a7 : SUCCESS in 7m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/60fb34d3857544be880e6d933cec7650 : SUCCESS in 11m 21s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b07dc4addb334a62b8b0538eea09e5f6 : SUCCESS in 13m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/22bd9a544825439eaefa5b424d35c9b5 : SUCCESS in 13m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/771e1c48c34f4de3b12b63fba8b74c38 : SUCCESS in 48m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/32a5869846974e7c97bf34219222ac76 : SUCCESS in 1h 14m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1909c202feb4d338d071b47031a8140 : SUCCESS in 1h 25m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/9ce4b930c49b4aa6907ec7870f13a1c7 : SUCCESS in 1h 17m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f287794ba24720ee332506edb3693b1956529f30","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-26 20:58:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aede779446b776b3988c6f5d0457581a175ffe33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 20:58:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"322b521c76a13b897ab4b5de9797f29c67ce0408","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 23:23:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee1f91443e024d9080f8af6292da273a : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06edbc6eda3a40d6ab5970965b616a9f : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8ff1e8711024e82b1497708fda1196c : SUCCESS in 8m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae84ce892d0f4976886d5ad6a95d564a : SUCCESS in 7m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6cfbfc94684c46d0945573937eaf0252 : SUCCESS in 8m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/297a64eb22f04761858db59476ceb380 : SUCCESS in 11m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/80978142d3564fcd9c2e21a6ce0e0209 : SUCCESS in 11m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/194e09adffdd4042909e099c159583b4 : SUCCESS in 7m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a040ecddc945482da278828926ba1bdf : SUCCESS in 11m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/823cf9db5f4c4335916ccbc6e7caa857 : SUCCESS in 10m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fa36374bff5f48829b0cdbd425b47c3c : SUCCESS in 17m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/25170d8331984c0dbcd6ee59247db226 : SUCCESS in 30m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0d293e8fc7a043eb8d74bfeba8702817 : SUCCESS in 16m 14s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4b261c5cfd7f4768ad383ddf3c37d4c6 : SUCCESS in 20m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a326f9b7f17b4e0b90fb1c6ae8436a9b : SUCCESS in 19m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/86957bad18a0400c912b2cb0c58dabd8 : SUCCESS in 7m 13s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7cd065f3245848a18fc41b8d9fac1f02 : SUCCESS in 7m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d54c489b7ead4b06b642f001553c2383 : SUCCESS in 8m 23s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f0703fc5ef304739a19744fff9d0213a : SUCCESS in 12m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1d0df9b3a00c42ae84af2c921c3ed517 : SUCCESS in 13m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3247838b7e34f068e140832d47ecd81 : SUCCESS in 1h 01m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/05c55b9877644124adcae393d4c15286 : SUCCESS in 1h 12m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/297cc19690d042e78114267407773429 : SUCCESS in 1h 24m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/4f958da0fef941d5ba3fbfbe7dd5089a : SUCCESS in 1h 21m 23s","accounts_in_message":[],"_revision_number":1},{"id":"33766df3baaf8e60ffe8fbad0fbef47a896a38ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 23:23:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da9740ad86c4c9a9b8d7a12d0d446febb15c2b34","revisions":{"da9740ad86c4c9a9b8d7a12d0d446febb15c2b34":{"kind":"REWORK","_number":1,"created":"2020-05-26 15:20:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/730850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/730850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/730850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/730850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/730850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/730850/1"}}},"commit":{"parents":[{"commit":"7d26c6599bb658e714e10f62b9fd86e5e1487b4a","subject":"update constraint for os-win to new release 5.0.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7d26c6599bb658e714e10f62b9fd86e5e1487b4a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-26 15:20:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-26 15:20:50.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 12.4.0","message":"update constraint for tripleo-common to new release 12.4.0\n\nChange-Id: I16a4438d059014e70273c59de0909a8d24155f88\nmeta:version: 12.4.0\nmeta:diff-start: -\nmeta:series: ussuri\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Commit: Wes Hayutin \u003cweshayutin@gmail.com\u003e\nmeta:release:Change-Id: I4c8ec629a8cc6ab72f7d6ba140ae71152f6d399e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: wes hayutin \u003cwhayutin@redhat.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+1: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da9740ad86c4c9a9b8d7a12d0d446febb15c2b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da9740ad86c4c9a9b8d7a12d0d446febb15c2b34"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
