)]}'
{"id":"openstack%2Frequirements~731152","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Ifc5af328544ebe35dc7f465f4fa426115235387e","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Ifc5af328544ebe35dc7f465f4fa426115235387e","subject":"update constraint for neutron to new release 15.1.0","status":"MERGED","created":"2020-05-27 10:40:28.000000000","updated":"2020-05-28 00:27:27.000000000","submitted":"2020-05-28 00:27:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731152-1590625647874-649d007f","meta_rev_id":"a76b2898c4c3e476fba27bb78273648e0a1988a9","_number":731152,"virtual_id_number":731152,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 00:27:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-05-27 14:25:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-05-27 21:14:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-05-27 21:14:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 14:25:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-27 21:14:12.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-28 00:27:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3eae56645fafb928badfdf2b615cb5bc8b10224e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-27 10:40:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44646dc2fe42dd7d4d3e9ba179f569562cfecdaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 12:50:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d29c91d28684624b0ce05850ff4b6f0 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d161a1a08cf44f5ea4697a7765456adc : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ccc365f41a9c47c2919c7919b69f2bbd : SUCCESS in 6m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be6baab4089c4401a0d0e625721c866b : SUCCESS in 5m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3e0b68c255d54ef292b9f9a5b4e4dfc1 : SUCCESS in 6m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7bcc94f37a5e459a81ad3b91d55e12be : SUCCESS in 5m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/8fbfd418fa5a407d9b02376ef65c25e8 : SUCCESS in 11m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1a068c7f819b45d6adac3c3ee532eb36 : SUCCESS in 9m 10s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b2a23fa7022e46a8859c6f43dc2dfeed : SUCCESS in 10m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d608da7be64245a6a7b8b5dd969a5b8e : SUCCESS in 11m 36s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4de8656d479d48088980f0d5191f7e2e : SUCCESS in 6m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ea1ae60c7c5c4dd690ea43deae49a770 : SUCCESS in 8m 14s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2f3dbcd93fa147cf8c58fe31676326c3 : SUCCESS in 11m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d879015c28c54cac9d4a5179d1e54da3 : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cc4c94f7ef12441f90a9d6681474121f : SUCCESS in 10m 49s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e498f7cad3164d36875872961e4ed001 : SUCCESS in 24m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d543be5c953f4d388c58724f4e711453 : SUCCESS in 19m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/9f9099fc2ce64fee84e8ea97b00b2467 : SUCCESS in 25m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/67febb0fdabf423d9920e2c1c5d6ea97 : SUCCESS in 24m 19s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/c3442def226a477cae7c5fc815552c36 : SUCCESS in 13m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4cefdb0a628c4865876c7820eb6adfd6 : SUCCESS in 13m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4655589397a14354bcd390ab6b08afeb : SUCCESS in 14m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7f97769518c14e20b7dc68e615e5e69f : SUCCESS in 19m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/67cf6ae052284f57b3d299ae60d14356 : SUCCESS in 7m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f2cea31beea84688831acab0aea2dd61 : SUCCESS in 9m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e3a9371476214359a04ae6b968f1fdfd : SUCCESS in 9m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b3aa11de37e49ec988e4e96ecba0eb2 : SUCCESS in 1h 13m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c50cc3e87fa345d3897a430b1417717b : SUCCESS in 1h 43m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/54719df724954b699e680e6c1a88c543 : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7d3e2093d18d7d9ad6f305c0c56a575ead231b92","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-27 14:25:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"96120a4dbd87067b0519befd49d445622b5b20bb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-27 21:14:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ab8106c3b8a29bbd5fe56ef9cb5abf4b051d473a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 21:14:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"712745d4f1959e2fbec1e0d7d4e3cc1da6f7ea33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 00:27:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ed0b252873a9139a7ec5b6cf2851fda9b3c85d22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 00:27:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/940f35d87441491a97efeaa04f309b58 : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e47a805dda404773bd3b5f3c76707514 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d370534908ec41faa845caad75b9ccf0 : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c3a5506055654caba8e2c96041834a38 : SUCCESS in 5m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a559be82dd34bc5959156af60960a03 : SUCCESS in 6m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d50f3f346ebe44cb84a396950941e9a1 : SUCCESS in 8m 18s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/718fab4c2635434584170fd2f7d88936 : SUCCESS in 11m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/af7fdcd1b0de4cc9bad7fe9f68e55384 : SUCCESS in 11m 56s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/73d35fe8ee4e4f70843e0abd20de7e4d : SUCCESS in 11m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5c9a059d946e4b7aa150c28c711220b9 : SUCCESS in 11m 20s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1dbf94fb5d284e939ee99f8a9659d8b3 : SUCCESS in 9m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/63ce38b7831041469ca5f46a1dee9edd : SUCCESS in 9m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/2cc212844f8747b8bc00559865293bb4 : SUCCESS in 13m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/09dcfca20c3e443ca00bf9be3600f6a9 : SUCCESS in 13m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/56a51a3961ac42679cbd14e264a1d75a : SUCCESS in 11m 56s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1f8264b89a5a4f3e89535e24c9538731 : SUCCESS in 20m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ece9d349b71f45dd930cc1677974f7b7 : SUCCESS in 23m 15s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c65bdee0776342de818eb768484d2c50 : SUCCESS in 28m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8eb210bd656a4669a38349ec2d74714f : SUCCESS in 22m 07s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f342825529a14a788c7bcdbea62948d9 : SUCCESS in 20m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a6193b497320463fb219382ab3dae240 : SUCCESS in 18m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/14f68611bc5d4576af604c2b156a01a6 : SUCCESS in 19m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/74c551c633ca41e2aaab1df1a6b145d1 : SUCCESS in 22m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f893f13ae8c642b3bd30ae9a96884841 : SUCCESS in 9m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/fd28b092cd794539bdca80e5ae708da6 : SUCCESS in 14m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/65ba4bbf27e941f28f4d82f399bfea0e : SUCCESS in 12m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/716d6c341de1416794aa59a2faef8658 : SUCCESS in 1h 13m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/50c1c51d2be143f79f2ea89b14ef64a3 : SUCCESS in 1h 36m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/530e0ac410184620abe8e6193bddd505 : SUCCESS in 1h 00m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58b6747152ca45a4e5c6dfa176f8e00cee6ff368","revisions":{"58b6747152ca45a4e5c6dfa176f8e00cee6ff368":{"kind":"REWORK","_number":1,"created":"2020-05-27 10:40:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/731152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/731152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/731152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/731152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/731152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/731152/1"}}},"commit":{"parents":[{"commit":"2bcbcbeccb0c039981b6fc107dafc2b60acf479f","subject":"update constraint for ovsdbapp to new release 0.17.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bcbcbeccb0c039981b6fc107dafc2b60acf479f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-27 10:40:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-27 10:40:27.000000000","tz":0},"subject":"update constraint for neutron to new release 15.1.0","message":"update constraint for neutron to new release 15.1.0\n\nChange-Id: Ifc5af328544ebe35dc7f465f4fa426115235387e\nmeta:version: 15.1.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Commit: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta:release:Change-Id: I859b67b0a1d1c60e24ec70ca0d924bb2c4a9736e\nmeta:release:Code-Review+1: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta:release:Code-Review+1: Brian Haley \u003chaleyb.dev@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58b6747152ca45a4e5c6dfa176f8e00cee6ff368"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58b6747152ca45a4e5c6dfa176f8e00cee6ff368"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
