)]}'
{"id":"openstack%2Frequirements~731381","triplet_id":"openstack%2Frequirements~master~I9ae13fdb702dd2f140bebd7db326b3b47ef7f284","project":"openstack/requirements","branch":"master","topic":"arm64-wheel","hashtags":[],"change_id":"I9ae13fdb702dd2f140bebd7db326b3b47ef7f284","subject":"Add more wheel builds to check","status":"MERGED","created":"2020-05-28 01:33:42.000000000","updated":"2020-05-29 01:47:00.000000000","submitted":"2020-05-29 01:06:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731381-1590714380169-41f9ad85","meta_rev_id":"759f556618c86eb055c86d75c874eeb3e2ed1eb4","_number":731381,"virtual_id_number":731381,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-29 01:06:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-28 22:33:16.000000000","_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-28 12:25:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-28 15:48:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-28 15:48:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2020-05-28 08:06:11.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 07:35:07.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-05-28 08:06:11.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-28 12:25:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-28 15:48:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-29 01:06:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7f282f10ff2f5ca7736f9e3212fadd7368851a40","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-28 01:33:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58279b424c9b29dc581ae8a7e2718e5dbae91324","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 01:56:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- build-wheel-mirror-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/3618e0e9f3b1495399e9cae76887a708 : RETRY_LIMIT in 2m 22s\n- build-wheel-mirror-ubuntu-xenial-arm64 https://zuul.opendev.org/t/openstack/build/9e4be2ecd2b14a35b03cf6c1ef4238c2 : FAILURE in 6m 36s\n- build-wheel-mirror-ubuntu-bionic-arm64 https://zuul.opendev.org/t/openstack/build/934fa46e2d0242e7845ac3a4149ec4e4 : FAILURE in 3m 21s\n- build-wheel-mirror-ubuntu-focal-arm64 https://zuul.opendev.org/t/openstack/build/1acb04f5c36e40008dbf31344b3a02fc : RETRY_LIMIT in 1m 50s","accounts_in_message":[],"_revision_number":1},{"id":"4b724afaae44396effcfde2228e2c43244f51212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 02:57:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73c2c6e2508b4c9f8ad67b7e3e7fcceb : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff5fddfb624f4b62a3f3c341577f1cd7 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4229ba9749d64cd69cde1a01faf82310 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f52b8e119a394e7d9159280e5d47f0da : SUCCESS in 4m 26s\n- build-wheel-mirror-centos-8 https://zuul.opendev.org/t/openstack/build/64d79d5e51ef42e29886edb28239a75b : RETRY_LIMIT in 2m 32s\n- build-wheel-mirror-ubuntu-focal https://zuul.opendev.org/t/openstack/build/463a9f30a89e4656a647f8e46a2840a2 : RETRY_LIMIT in 1m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/608a03d82f5a489b866211d58d0fb12e : SUCCESS in 3m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7741df034bb3417fa2b0b8d2b01c6296 : SUCCESS in 51m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/caa2e1dd57fa4c09b67460c816c82867 : SUCCESS in 1h 10m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a95e41b03144f07b7fd629edae20e1b : SUCCESS in 1h 22m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/1c38056c9a88457b8eb7269fa1cc5eb1 : SUCCESS in 1h 06m 33s","accounts_in_message":[],"_revision_number":1},{"id":"833400c50e9cbb576809d4f92962e685c7e6a7dc","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-28 03:05:59.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"58e58ecd98e4773ecae40025b506703fc756a216","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 03:23:20.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- build-wheel-mirror-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/a53eb95638ce42c38b76e0ae7286e0f2 : RETRY_LIMIT in 2m 15s\n- build-wheel-mirror-ubuntu-xenial-arm64 https://zuul.opendev.org/t/openstack/build/4bce613e0b0c43b687640aaae5ff205d : FAILURE in 3m 26s\n- build-wheel-mirror-ubuntu-bionic-arm64 https://zuul.opendev.org/t/openstack/build/f8b97d2bc51442ceb1365becc0f6f77f : FAILURE in 3m 28s\n- build-wheel-mirror-ubuntu-focal-arm64 https://zuul.opendev.org/t/openstack/build/1c229fbf31cd4608b5d9ab9fea0b1ab1 : FAILURE in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"75ba1d27952812d31e4437e81b0c329369246859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 04:27:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/075ca7d7d76f402e9106aa04dbd38b09 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce2f28d9c02c435b8371600a85cec245 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4696b58c767402793ca66f1113c9ffb : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f194c2fd84bc4272a21a466761bbd74b : SUCCESS in 7m 00s\n- build-wheel-mirror-centos-8 https://zuul.opendev.org/t/openstack/build/31bc05e8447447d29c4969d19d67ee90 : RETRY_LIMIT in 2m 04s\n- build-wheel-mirror-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6f0afe53179471f91dfdcf8d70be665 : FAILURE in 6m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c47a627f774457fbdf96ed03c7642bd : SUCCESS in 5m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb5645e601434a349f1eee07714854fb : SUCCESS in 57m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5cbedbb2c044d3f9fb7797ed0c1b3fd : SUCCESS in 1h 16m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a225be2c986149abb612072e09cf4295 : SUCCESS in 53m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/8f38876901924ff99ae769b159dd96ff : SUCCESS in 54m 55s","accounts_in_message":[],"_revision_number":2},{"id":"82907689124756dfb1333704c8e41fb53d88142a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-28 06:15:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"daa9bc644c50db1031f9e21f42d1e3f3f02ed5b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 06:16:42.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/requirements on branch master.  The error was:\n\n  while scanning a plain scalar\n  in \"openstack/requirements/.zuul.d/project.yaml@master\", line 72, column 17\nfound a tab character that violate indentation\n  in \"openstack/requirements/.zuul.d/project.yaml@master\", line 73, column 1","accounts_in_message":[],"_revision_number":3},{"id":"ea040b9215c01671c588538e5c82c52ea8962ac2","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-28 06:18:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"690a19bfb31456a2ff440a5166314c11f43e86f5","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-28 06:20:17.000000000","message":"Topic changed from arm64-wheels to arm64-wheel","accounts_in_message":[],"_revision_number":4},{"id":"4c2edd3546b912e45b7d87a7901128667d61f4f5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-28 07:32:28.000000000","message":"Patch Set 4:\n\nIMHO all those jobs should run on either bindep.txt change or openstack/requirements/upper-constraints.txt change.\n\nSo whenever new dependency comes it gets build.","accounts_in_message":[],"_revision_number":4},{"id":"2291ae59b473237f4b464f1f14709c382da84bf0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-28 07:35:07.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"950fc43593fa6a42defda923d4453728ddaa73dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 07:48:19.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"73351f29fc646c730a23a391197c288ed73b23cf","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-28 08:06:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b60a4bc1866e003fe0f841d1fdbfa65a590cb941","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 09:29:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8af3ee6436e94ee39049aeba34b1e3c0 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad270e49a30546288485a03a9b7c4518 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e0360b6e3a0409ba335331fcd420fe2 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ebfb2d8b00d4c73867b8f5cdbce129f : SUCCESS in 5m 11s\n- build-wheel-cache-centos-8 https://zuul.opendev.org/t/openstack/build/01566169cb6e4d5ebb9eb62f157faafd : SUCCESS in 1h 10m 00s\n- build-wheel-cache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28d0f6d982a74652bd7d9ba42b6eb6c3 : SUCCESS in 1h 16m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2e0344030c146faaac0feef72bac596 : SUCCESS in 3m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5756e017b5d2434f9ca043607370378e : SUCCESS in 1h 05m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce2b894980d54745ab040fef88db9fa9 : SUCCESS in 1h 10m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30a2bca327494f628c24a1fe5f3ba462 : SUCCESS in 1h 20m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/d8bbb790e56f46349a9debdd95ca255e : SUCCESS in 1h 11m 57s","accounts_in_message":[],"_revision_number":5},{"id":"3107979e3e755a44ce7506b0e96d50b4bc80806b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 10:31:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- build-wheel-cache-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/4a1b01e715c7496096d3ef5666cb3144 : SUCCESS in 1h 55m 01s\n- build-wheel-cache-ubuntu-xenial-arm64 https://zuul.opendev.org/t/openstack/build/95a67e8605114cc8b4b2c8fc9d2b712b : SUCCESS in 2h 12m 13s\n- build-wheel-cache-ubuntu-bionic-arm64 https://zuul.opendev.org/t/openstack/build/e67578d15ed4481fba4f5d2bce18e7a0 : SUCCESS in 2h 09m 34s\n- build-wheel-cache-ubuntu-focal-arm64 https://zuul.opendev.org/t/openstack/build/8349b8416eb746e6a7a216114f640304 : SUCCESS in 2h 23m 30s","accounts_in_message":[],"_revision_number":5},{"id":"467eb8096dbe22db8ff3eca48d419187873dad7e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-28 12:25:30.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"6b79183a2e37c9293825aa44a0185d8108097996","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-28 15:48:11.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"bd97885b405e5d444fc1738c9a2c63a4fbd230ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 16:37:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"587f82835b7eddf920084530179280fd054da265","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 21:30:55.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d40c0680f7f94047a204ff3fb3f38113 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3106c37849ae4368901a767426c99464 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a91d956ad7b5443cbf0034daa885f0dd : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca74b99bbf8748e28e70d7a729cb18f2 : SUCCESS in 7m 27s\n- build-wheel-cache-centos-8 https://zuul.opendev.org/t/openstack/build/42e78085d62b4dc9aa38f758b742fcc4 : SUCCESS in 59m 54s\n- build-wheel-cache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c735f0b59d234c6498b388c2d367f87b : TIMED_OUT in 2h 40m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/02b0d38389ba46a4b85c9d0f90ef5fd5 : SUCCESS in 6m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/785598732c514ffe8da994a38e4063d8 : SUCCESS in 1h 14m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/80d4485fbea44bc5b4fe2d1bc47c0677 : SUCCESS in 1h 27m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3dceb011a3894daca8b8af4e9936ff82 : SUCCESS in 1h 34m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/d2f6d2428dff4fdf847104e0d777839e : SUCCESS in 1h 19m 49s","accounts_in_message":[],"_revision_number":5},{"id":"121444d3aa3c295e9bd2e28fd830e5a780f99319","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-28 22:33:16.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"36c4b0ced62328ba6cf4c608169d482a72fc0710","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 23:54:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5574f83260f144beb149320c05dc7f8f : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4f272d946e947849f736d97937cfa2b : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d760588d346844cba93c200d29de29ba : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2216b0d305f947248ba64a7464eb9e3b : SUCCESS in 6m 43s\n- build-wheel-cache-centos-8 https://zuul.opendev.org/t/openstack/build/8bf95739936d43d4a41c54406e071e1c : SUCCESS in 1h 00m 14s\n- build-wheel-cache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da270898a7d14408898585d2b1a3b6e3 : SUCCESS in 1h 09m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae65ecb492b940cbbc66fc85f445d016 : SUCCESS in 5m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8bfa96fc5694955a2f10313c7307725 : SUCCESS in 55m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9d1027236fd1432f87eaf2420ad0684b : SUCCESS in 1h 08m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8659fbb56a624c42afcef353f57eaf59 : SUCCESS in 1h 20m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/e96909910e024058a15c9126481d02f6 : SUCCESS in 55m 46s","accounts_in_message":[],"_revision_number":5},{"id":"315c1c416654ee90fc3f9ed2ca393d9bf16324c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 23:55:11.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"27c19d791e913b9ae210b0727a5c64b56589f984","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 01:03:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- build-wheel-cache-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/a57f61413b854775829fc85c4baac729 : SUCCESS in 2h 01m 33s\n- build-wheel-cache-ubuntu-xenial-arm64 https://zuul.opendev.org/t/openstack/build/2afbb4471c3d4cf78b66eafccecd899b : SUCCESS in 2h 07m 33s\n- build-wheel-cache-ubuntu-bionic-arm64 https://zuul.opendev.org/t/openstack/build/d581f0361811445cb63d44521d551106 : SUCCESS in 2h 03m 23s\n- build-wheel-cache-ubuntu-focal-arm64 https://zuul.opendev.org/t/openstack/build/d5ce75f877e34f3e81cae121e25b5220 : SUCCESS in 2h 24m 27s","accounts_in_message":[],"_revision_number":5},{"id":"a5e46ed61ef986185e9b3d9f4633bf325b37a008","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 01:06:19.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a4beec8aa174e158757b29edfe29e9f : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb7a84b590094445bd39640b67c571b4 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cd69f823c1748c5b234202828bd6243 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7296a670b0247b9afd34e70ace7af6f : SUCCESS in 4m 53s\n- build-wheel-cache-centos-8 https://zuul.opendev.org/t/openstack/build/9dff3712b84f458b96739f1a6852c6bb : SUCCESS in 55m 23s\n- build-wheel-cache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ac3022215de4b82858961efd96e6562 : SUCCESS in 47m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d2d1c61ea45434f9bfcf9eb54f11589 : SUCCESS in 3m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12ddb96241ee43d7837917d41c972890 : SUCCESS in 38m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7aeb7f90aa824b9c96d0ee09d310fa59 : SUCCESS in 1h 08m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/822d57eaec474cd48211132bb0afcc6b : SUCCESS in 1h 00m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/de267504d44d4ae5bac3f4d93a4c02f0 : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":5},{"id":"e9e8f1c09de0b7854b94896a4977b866ddffc3ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 01:06:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"0c77e43d366519019a7ef4bc2a87088e3f7d183c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 01:09:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f742fad056454ff7ac81755185a8d516 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":5},{"id":"759f556618c86eb055c86d75c874eeb3e2ed1eb4","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-29 01:47:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f565d76fdb1d107b9a82dee538eefa28c82be8b6","revisions":{"0c95b405852df5bf8a1d087120b64d4a15a32221":{"kind":"REWORK","_number":1,"created":"2020-05-28 01:33:42.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/81/731381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/731381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/731381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/731381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/731381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/731381/1"}}},"commit":{"parents":[{"commit":"cf74d072c7b6dae3a2de272820b645062529b1a6","subject":"Fix \u0027oslo.i18n\u0027 requirement error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf74d072c7b6dae3a2de272820b645062529b1a6"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 01:29:46.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 01:29:46.000000000","tz":600},"subject":"Add more wheel builds to check","message":"Add more wheel builds to check\n\nThis adds centos-8 and ubuntu-focal to the x86 wheel builds.\n\nIt also adds the arm64 builds in the check-arm64 queue.  This is\ncurrently a non-voting queue -- however failures here will actually\nstop the periodic wheel release job.  There might be an argument for\nmoving them into check/gate; we have been a little worried that\ninsufficient arm64 resources would hold up the gate for changes that\nrequired timely merge; however the linaro cloud is certainly seeming\nto be stable and performant.  We can start here as we initially debug.\n\nChange-Id: I9ae13fdb702dd2f140bebd7db326b3b47ef7f284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c95b405852df5bf8a1d087120b64d4a15a32221"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0c95b405852df5bf8a1d087120b64d4a15a32221"}]},"branch":"refs/heads/master"},"104fe759733bcc78ca5f61b985e4b9f13e3d9e78":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-28 03:05:59.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/81/731381/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/731381/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/731381/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/731381/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/731381/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/731381/2"}}},"commit":{"parents":[{"commit":"6ac56aecfbce89f3711cc0d0d5a5fb95a98f3737","subject":"Extend libzmq negative match to focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ac56aecfbce89f3711cc0d0d5a5fb95a98f3737"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 01:29:46.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 02:52:39.000000000","tz":600},"subject":"Add more wheel builds to check","message":"Add more wheel builds to check\n\nThis adds centos-8 and ubuntu-focal to the x86 wheel builds.\n\nIt also adds the arm64 builds in the check-arm64 queue.  This is\ncurrently a non-voting queue -- however failures here will actually\nstop the periodic wheel release job.  There might be an argument for\nmoving them into check/gate; we have been a little worried that\ninsufficient arm64 resources would hold up the gate for changes that\nrequired timely merge; however the linaro cloud is certainly seeming\nto be stable and performant.  We can start here as we initially debug.\n\nChange-Id: I9ae13fdb702dd2f140bebd7db326b3b47ef7f284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/104fe759733bcc78ca5f61b985e4b9f13e3d9e78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/104fe759733bcc78ca5f61b985e4b9f13e3d9e78"}]},"branch":"refs/heads/master"},"a3be89d46b57dca07cc864b63528724779c80f6b":{"kind":"REWORK","_number":3,"created":"2020-05-28 06:15:30.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/81/731381/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/731381/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/731381/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/731381/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/731381/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/731381/3"}}},"commit":{"parents":[{"commit":"f0af02f6f3c499f061591b25a779d38280b2cea2","subject":"bindep.txt: Install python3-devel for CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0af02f6f3c499f061591b25a779d38280b2cea2"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 01:29:46.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 06:14:57.000000000","tz":600},"subject":"Add more wheel builds to check","message":"Add more wheel builds to check\n\nThis adds centos-8 and ubuntu-focal to the x86 wheel builds.\n\nIt also adds the arm64 builds in the check-arm64 queue.  This is\ncurrently a non-voting queue -- however failures here will actually\nstop the periodic wheel release job.  There might be an argument for\nmoving them into check/gate; we have been a little worried that\ninsufficient arm64 resources would hold up the gate for changes that\nrequired timely merge; however the linaro cloud is certainly seeming\nto be stable and performant.  These do not work right now, and are\nfixed up in follow-ons.  We can start here as we initially debug.\n\nChange-Id: I9ae13fdb702dd2f140bebd7db326b3b47ef7f284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3be89d46b57dca07cc864b63528724779c80f6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a3be89d46b57dca07cc864b63528724779c80f6b"}]},"branch":"refs/heads/master"},"278eafd82e32d72a6beec3311d30776244d1d61b":{"kind":"REWORK","_number":4,"created":"2020-05-28 06:18:50.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/81/731381/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/731381/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/731381/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/731381/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/731381/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/731381/4"}}},"commit":{"parents":[{"commit":"f0af02f6f3c499f061591b25a779d38280b2cea2","subject":"bindep.txt: Install python3-devel for CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f0af02f6f3c499f061591b25a779d38280b2cea2"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 01:29:46.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 06:18:20.000000000","tz":600},"subject":"Add more wheel builds to check","message":"Add more wheel builds to check\n\nThis adds centos-8 and ubuntu-focal to the x86 wheel builds.\n\nIt also adds the arm64 builds in the check-arm64 queue.  This is\ncurrently a non-voting queue -- however failures here will actually\nstop the periodic wheel release job.  There might be an argument for\nmoving them into check/gate; we have been a little worried that\ninsufficient arm64 resources would hold up the gate for changes that\nrequired timely merge; however the linaro cloud is certainly seeming\nto be stable and performant.  These do not work right now, and are\nfixed up in follow-ons.  We can start here as we initially debug.\n\nChange-Id: I9ae13fdb702dd2f140bebd7db326b3b47ef7f284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/278eafd82e32d72a6beec3311d30776244d1d61b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/278eafd82e32d72a6beec3311d30776244d1d61b"}]},"branch":"refs/heads/master"},"f565d76fdb1d107b9a82dee538eefa28c82be8b6":{"kind":"REWORK","_number":5,"created":"2020-05-28 08:06:11.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/81/731381/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/731381/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/731381/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/731381/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/731381/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/731381/5"}}},"commit":{"parents":[{"commit":"50c098d42b0197ab016039072559ef5240ac8da7","subject":"bindep.txt: Install python3-devel for CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50c098d42b0197ab016039072559ef5240ac8da7"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 01:29:46.000000000","tz":600},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-28 08:05:43.000000000","tz":120},"subject":"Add more wheel builds to check","message":"Add more wheel builds to check\n\nThis adds centos-8 and ubuntu-focal to the x86 wheel builds.\n\nIt also adds the arm64 builds in the check-arm64 queue.  This is\ncurrently a non-voting queue -- however failures here will actually\nstop the periodic wheel release job.  There might be an argument for\nmoving them into check/gate; we have been a little worried that\ninsufficient arm64 resources would hold up the gate for changes that\nrequired timely merge; however the linaro cloud is certainly seeming\nto be stable and performant.  These do not work right now, and are\nfixed up in follow-ons.  We can start here as we initially debug.\n\nChange-Id: I9ae13fdb702dd2f140bebd7db326b3b47ef7f284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f565d76fdb1d107b9a82dee538eefa28c82be8b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f565d76fdb1d107b9a82dee538eefa28c82be8b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
