)]}'
{"id":"openstack%2Frequirements~731394","triplet_id":"openstack%2Frequirements~master~I171c94b56d1243efb83e5b457d45f23e21b3be10","project":"openstack/requirements","branch":"master","topic":"arm64-wheel","hashtags":[],"change_id":"I171c94b56d1243efb83e5b457d45f23e21b3be10","subject":"Move wheel build jobs out of project-config","status":"MERGED","created":"2020-05-28 03:05:59.000000000","updated":"2020-05-28 20:42:23.000000000","submitted":"2020-05-28 20:38:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731394-1590698336265-41f9ad85","meta_rev_id":"cc938cfeac8fb151bdf313a2cf33f3d891813915","_number":731394,"virtual_id_number":731394,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 20:38:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-28 12:26:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-28 15:46:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-05-28 07:54:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-05-28 07:35:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-05-28 16:36:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-28 16:37:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 07:35:03.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-05-28 07:54:15.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-28 16:36:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-28 16:37:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-28 20:38:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08aa318f233de68afb58b8f90ba921e180b1d7cc","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-28 03:05:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"837efd3c2136ab27acaa2fa14848e33cc47bd614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 03:22:07.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"d0015e8ccb787c53478c10d3104477d1e1368e83","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-28 03:40:36.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e6fd0702c95edd19ad58dcfdacb412ec38c827f9","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-28 04:00:19.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"daa3e54c986518c33f1387d7390bbf6520a576b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 05:11:28.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"a2c0981bb52cb9f23eb34f0c7cd8ea00da4b58f6","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-28 06:15:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"90731c1a47d9946a2b16c2880906c5f4fd77817c","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-28 06:24:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f445cbba2edc804716d5e3903ae5a59609a6e35a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-28 07:35:03.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"eb75a0fac042fea5116f94df54d0db5df12d0f69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 07:48:20.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"091bbb812c19919564b76dac1aad847f52cc0281","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-28 07:52:47.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"1d53572a0f19b3e2e0458092fb28bc33694ddfb4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-28 07:54:15.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"4668829af1e3976c3abc2486d136bd03a02e8a2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 09:22:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/749aed31eee54288b8d171e1ac83200b : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1a2fe18b00f4a64a7bcab78258705e1 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1057aa76e8a046f6971d98af6b418119 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/215175c8c71449aeb2ea6915f000b5b4 : SUCCESS in 4m 41s\n- build-wheel-cache-centos-7 https://zuul.opendev.org/t/openstack/build/6b8c2fbc23c347aaac5da531cede7b25 : SUCCESS in 55m 32s\n- build-wheel-cache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3edf0802d7f43dca3987859406204de : SUCCESS in 53m 00s\n- build-wheel-cache-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/af971cfa458a4803aaeba6ad80661316 : SUCCESS in 51m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/13f79cb34f8f4658abe880f812df10da : SUCCESS in 5m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8737281d336d42ceb430bb02f20d114c : SUCCESS in 59m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2ebd71cb343e4e1ebdc2474aaeebbfe5 : SUCCESS in 1h 10m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b34d5e33f7cf4b7e835f47e649697ccc : SUCCESS in 1h 27m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/5ca2c131c2d24e05a3f58c0e1b900902 : SUCCESS in 1h 11m 08s","accounts_in_message":[],"_revision_number":5},{"id":"c1d71c6ed561ceaef62ccff191145d3786cabe35","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-28 12:26:37.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7d97216ae00ad133f6e6553a72cb86e47dbc0d98","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-28 15:46:19.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c91c1bfd59c5d0c354c73c195dd69370adfd26b2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-28 16:36:41.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a9d8112ba03a539f42c3d15a048a131cfc5e7e90","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-28 16:37:06.000000000","message":"Patch Set 5: -Workflow","accounts_in_message":[],"_revision_number":5},{"id":"1fa70eb22ed8d5c6dbf425754292265d7b59dd72","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-28 16:37:15.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"fc505d3f3f1ddcbc9049bd49b97fcafb80ea09e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 16:37:24.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ec6e4c8ecb1bc3945806ac64fb8cdb2a85a2beae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 20:38:55.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a95f5cc207224dd48b2e7ec7fbfa064e : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d75a127ad83462984c723b26b5edd5e : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a2c7672b8db4b6f91ed33e2da8e9047 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31362354d4a64895accb7299654b40ea : SUCCESS in 4m 27s\n- build-wheel-cache-centos-7 https://zuul.opendev.org/t/openstack/build/631f86520be54460bd1dba29989f3e45 : SUCCESS in 58m 13s\n- build-wheel-cache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d94d4572a50b427f8a35718322f974f0 : SUCCESS in 55m 36s\n- build-wheel-cache-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/d836cdc08663457e8bd9451e945bcfa2 : SUCCESS in 1h 01m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ef20c4bcc69414fa0f7b33fb6680e7a : SUCCESS in 3m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4871f33220534a778e7184dde369a8b8 : SUCCESS in 51m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/99ec15bb387b49398e562a8f8829a8c8 : SUCCESS in 1h 05m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d0d42ddbf69421ea7c5667ab7966517 : SUCCESS in 1h 33m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/5058dbe589734449aef9235c5716e017 : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":5},{"id":"e70671ba12fe0e47493ceec6fb0815087242ebcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 20:38:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"59c3d67b084bf617c517130f8e6e54ab3d83436d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 20:42:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/291b810ff84847a8bcec7f16d89a10d8 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d21518b0e3f766afc5b4912524b38e28b589557e","revisions":{"f01be84975b93ea2678f2c9fba7c8581393fa7fc":{"kind":"REWORK","_number":1,"created":"2020-05-28 03:05:59.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/94/731394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/731394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/731394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/731394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/731394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/731394/1"}}},"commit":{"parents":[{"commit":"104fe759733bcc78ca5f61b985e4b9f13e3d9e78","subject":"Add more wheel builds to check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/104fe759733bcc78ca5f61b985e4b9f13e3d9e78"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 03:05:16.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 03:05:16.000000000","tz":600},"subject":"[wip] move wheel build jobs out of project-config","message":"[wip] move wheel build jobs out of project-config\n\nso we can specutalively test\nDepends-On: https://review.opendev.org/731392\n\nChange-Id: I171c94b56d1243efb83e5b457d45f23e21b3be10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f01be84975b93ea2678f2c9fba7c8581393fa7fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f01be84975b93ea2678f2c9fba7c8581393fa7fc"}]},"branch":"refs/heads/master"},"fbd76e2818a9d74bf0ef128ca903439c2ed68863":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-28 03:40:36.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/94/731394/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/731394/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/731394/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/731394/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/731394/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/731394/2"}}},"commit":{"parents":[{"commit":"104fe759733bcc78ca5f61b985e4b9f13e3d9e78","subject":"Add more wheel builds to check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/104fe759733bcc78ca5f61b985e4b9f13e3d9e78"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 03:05:16.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 03:38:01.000000000","tz":600},"subject":"[wip] move wheel build jobs out of project-config","message":"[wip] move wheel build jobs out of project-config\n\nso we can specutalively test\n\nDepends-On: https://review.opendev.org/731408\nChange-Id: I171c94b56d1243efb83e5b457d45f23e21b3be10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbd76e2818a9d74bf0ef128ca903439c2ed68863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fbd76e2818a9d74bf0ef128ca903439c2ed68863"}]},"branch":"refs/heads/master"},"e5562b2f019f9b8a63e0df95747529f65f6036a9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-05-28 04:00:19.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/94/731394/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/731394/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/731394/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/731394/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/731394/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/731394/3"}}},"commit":{"parents":[{"commit":"104fe759733bcc78ca5f61b985e4b9f13e3d9e78","subject":"Add more wheel builds to check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/104fe759733bcc78ca5f61b985e4b9f13e3d9e78"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 03:05:16.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 03:59:20.000000000","tz":600},"subject":"[wip] move wheel build jobs out of project-config","message":"[wip] move wheel build jobs out of project-config\n\nso we can specutalively test\n\nDepends-On: https://review.opendev.org/731425\nChange-Id: I171c94b56d1243efb83e5b457d45f23e21b3be10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e5562b2f019f9b8a63e0df95747529f65f6036a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e5562b2f019f9b8a63e0df95747529f65f6036a9"}]},"branch":"refs/heads/master"},"c0a9821a1baa7df3f33d3cda98f2b0ae3e131581":{"kind":"REWORK","_number":4,"created":"2020-05-28 06:15:30.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/94/731394/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/731394/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/731394/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/731394/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/731394/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/731394/4"}}},"commit":{"parents":[{"commit":"cf74d072c7b6dae3a2de272820b645062529b1a6","subject":"Fix \u0027oslo.i18n\u0027 requirement error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf74d072c7b6dae3a2de272820b645062529b1a6"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 03:05:16.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 06:14:06.000000000","tz":600},"subject":"Move wheel build jobs out of project-config","message":"Move wheel build jobs out of project-config\n\nThis moves the wheel build jobs into openstack-zuul-jobs so we can\nDepends-On changes and test here.\n\nNote this has a dependency on a series of changes to the wheel build\njobs to support focal and arm64, which are added in changes after\nthis.\n\nDepends-On: https://review.opendev.org/731425\nChange-Id: I171c94b56d1243efb83e5b457d45f23e21b3be10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0a9821a1baa7df3f33d3cda98f2b0ae3e131581"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0a9821a1baa7df3f33d3cda98f2b0ae3e131581"}]},"branch":"refs/heads/master"},"d21518b0e3f766afc5b4912524b38e28b589557e":{"kind":"REWORK","_number":5,"created":"2020-05-28 06:24:46.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/94/731394/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/731394/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/731394/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/731394/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/731394/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/731394/5"}}},"commit":{"parents":[{"commit":"cf74d072c7b6dae3a2de272820b645062529b1a6","subject":"Fix \u0027oslo.i18n\u0027 requirement error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cf74d072c7b6dae3a2de272820b645062529b1a6"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 03:05:16.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 06:22:44.000000000","tz":600},"subject":"Move wheel build jobs out of project-config","message":"Move wheel build jobs out of project-config\n\nThis moves the wheel build jobs into openstack-zuul-jobs so we can\nDepends-On changes and test here.\n\nNote this has a dependency on a series of changes to the wheel build\njobs to support focal and arm64, which are added in changes after\nthis.\n\nDepends-On: https://review.opendev.org/731425\nChange-Id: I171c94b56d1243efb83e5b457d45f23e21b3be10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d21518b0e3f766afc5b4912524b38e28b589557e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d21518b0e3f766afc5b4912524b38e28b589557e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
