)]}'
{"id":"openstack%2Frequirements~731515","triplet_id":"openstack%2Frequirements~master~I06f8ee408db610f4bd74c047de2f14b85b239362","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I06f8ee408db610f4bd74c047de2f14b85b239362","subject":"update constraint for python-magnumclient to new release 3.1.0","status":"MERGED","created":"2020-05-28 12:41:15.000000000","updated":"2020-05-30 19:18:59.000000000","submitted":"2020-05-30 19:17:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731515-1590866251091-8a0ad39b","meta_rev_id":"d673fb304872d91193907906c9c8abd2b0e4bc0d","_number":731515,"virtual_id_number":731515,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-30 19:17:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-28 12:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-28 15:53:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-28 15:53:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 12:48:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-28 15:53:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-30 19:17:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62e3cb2068d590ecb37e8a8aa846f9988a6dbf93","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-28 12:41:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f59a041ab9b585028b64607865277fc25676d28e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-28 12:48:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9a37411b7e27aa79b686a3fbdc509f090d540d43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 14:06:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f96f93f361f416a89451163d4fdead8 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b5077c932c94929b8d12abcc0c7247f : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40c732d69d9c43cebf1769f6e57d1a8e : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e553805c4b894f51a76a66925fd947f3 : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d87b8d28d4314df3904977d6f1fccf05 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2390b7e7c6444841bff9223242c3ef6d : SUCCESS in 5m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6cb918c819514d5f850ccc332a02b37e : SUCCESS in 9m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/44020fef3f984af9b0ba8061d01f66bc : SUCCESS in 10m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ef90dfe9ebd143078742feaf42d6b879 : SUCCESS in 8m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1e55fbdcc4b8457abb2d1179a3bd2ac6 : SUCCESS in 8m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5b6fb07a1035484d9d07d995ce5b6319 : SUCCESS in 13m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/778f66fed1f4404a9e865bc747572a2e : SUCCESS in 8m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/95c1e67a43084ba69846c50b5fa4e56a : SUCCESS in 15m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/08effd0e193849aba978dd2258d0be6d : SUCCESS in 38m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c8d07e2da72441958fbc887d3be8e469 : SUCCESS in 17m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9fc0c09d344e49b1b246174e30378e9a : SUCCESS in 20m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/046ffe79e151473c8cf6748752d51508 : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/62bbbc6911d94740ac49f3425106657b : SUCCESS in 9m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/876aee8b175442eea5c64c43d987ddc4 : SUCCESS in 14m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16b1f7e851334f5bacb36d9dbcaea164 : SUCCESS in 1h 03m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3049857a2ac14c90a775d042c3fce5cb : SUCCESS in 1h 15m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21c6d0010257424d8ad4800d9d224cd5 : SUCCESS in 1h 19m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/a91868e4bef44fe5bc8b9d7e9164fb5c : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c9ad102a4cf5b3ae642d4b0b71030a5172a29fc8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-28 15:53:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c3478b9c515af6f9a2284efb344ffaeb9508920e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 15:53:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed452ff661b0f81f6b1f198ef83ebd687db90b27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 17:11:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b7fc3e6bc8d44e296b64b6a5d012618 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e28b1fafc49a45758ae4bb58c4c5a977 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/314bc64686c241edb8eeee43e681b987 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e622c70e852a431cade2a4c98d2717eb : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/166cbeef080745a2a7137b03a755dc82 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/59a598192f784f7690831ee25075f03f : SUCCESS in 7m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c3b6f65fcccd410484f2041e1b6d8220 : SUCCESS in 15m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d3e7139613184742ae197f7cb365d124 : SUCCESS in 18m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2b21b5de32a44ce8bee9beec6c1fab9d : SUCCESS in 13m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8c39d20d202343ad9d78551ccfd40d89 : SUCCESS in 7m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/27aef4efd7794e389e7baae42c28ddda : SUCCESS in 12m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b353a670d0e4004af30b4af2d105012 : SUCCESS in 10m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/51006316662447d88e83b8a33fe0e391 : SUCCESS in 22m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/120b8c3d168c4f9bbd3416d8a144e132 : SUCCESS in 31m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fbc67297622643b39f4f83db4e5e08be : SUCCESS in 14m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1155ca87107b42d0b88e61136c02af47 : SUCCESS in 18m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f42f920aa5a64afd9553f7fef3b91f66 : SUCCESS in 6m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/201c845c7fa749a8a24b06ca6c0be0ff : SUCCESS in 8m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/a3288a7e200d4102b4b4f45d7352f2d8 : SUCCESS in 13m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3d8846d949bb4f2fb70098d499cc4097 : SUCCESS in 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91da2c51a980471391fc43291671d4b2 : SUCCESS in 52m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eadbbc43e941426a91f3fdee1bdc5fcc : SUCCESS in 1h 05m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9f4b2dcace04f88ad62c7f9921cfa96 : FAILURE in 1h 06m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/b17d6b2e01cf4840a4efc3cf4a5169eb : SUCCESS in 1h 03m 13s","accounts_in_message":[],"_revision_number":1},{"id":"074877c8263d773f4d852aa811c27b378d2ce423","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-28 17:23:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7a351aa69cc133f4aa66582417ef657f6b71942f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 18:56:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14209161354e46a28f2f89d2623cb93a : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/347b7e3f45c84b32b9c11fde7420a169 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bce96f237c14d8dac35990b0157081e : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ede1d3dbcdd949e7a82f6cfba758cfc1 : SUCCESS in 6m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9947903e7264ee9ab310fc34f440288 : SUCCESS in 5m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f9cb01a0ee849d2bf31825ec9fd7ce3 : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b845f4f2fca64fb2978fa56488bcdca8 : SUCCESS in 8m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5efaff54c22049a2a19046a97a2aac84 : SUCCESS in 10m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6b2e852dbcda458597c7f2c207740108 : SUCCESS in 12m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f6ad2003a12746af90917a87e54b0db7 : SUCCESS in 6m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5db5d7287c1f4862b69287940c509346 : SUCCESS in 12m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f995f194666348389c0aff5c27f52eb3 : SUCCESS in 8m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6b759738935a47d6b0ed30443b4ed48f : SUCCESS in 20m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d317a2961dee4d0881bd39393093f71f : SUCCESS in 17m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2914a0648b1b4a8bbec087775f0bd5a2 : SUCCESS in 18m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1176e5cd65464631bd696623a4721e38 : SUCCESS in 20m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b540141886274eeeb304a7a7c47ac7ec : SUCCESS in 7m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a9d123acc4a144c8b389be68b3599866 : SUCCESS in 8m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/33cd12ad6e704abbb442d772b15a402e : SUCCESS in 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/110d89d0b6534b3db888538d6b53d013 : SUCCESS in 49m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/be44a32b5ca142e4b7f255d712a350a8 : SUCCESS in 1h 08m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f3eb7b6044a499abdfb08510cde14a9 : SUCCESS in 1h 23m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/bee6fb6ee4d749829d947d66a206a53a : SUCCESS in 1h 07m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c480a6062549420a711839471f6ffafe2cd7d2db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 18:57:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41edcba7ea00108f3e0044096c49f0423d27b635","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-28 20:49:17.000000000","message":"Abandoned\n\nRandom gate failure.","accounts_in_message":[],"_revision_number":1},{"id":"6952638baaaa482acb9991b05f1e58a54293006c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-28 20:49:25.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"48b5cd0e5ee44ed65609e4993863757e4e07c90f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 22:17:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a5811115ed4420e8848603dd82fda5b : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa2a0ee476bd49a4828e662d8db7db01 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38249978ec6d480c8c144a1d863ce922 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68c4f3f2a4ae44dba4c9c3a74a04384c : SUCCESS in 7m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8f813ef3b7f490aab2ad68588324dce : SUCCESS in 6m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0ebdfce66efa4307ae2314e94f0f582a : SUCCESS in 6m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/56149aa222634deab70090bc984838ab : SUCCESS in 11m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/92eefad65e3b44b0a257b1b5104a5acc : SUCCESS in 13m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bd9dbffbdad242329d1607b5ceac59fd : SUCCESS in 10m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8a839586136747cb8949c1abeb6bb149 : SUCCESS in 5m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7295cc68910948b48a6fcd9d08a443a9 : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c816f7f5328b4cbda6ce66ab54d00c92 : SUCCESS in 8m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0c4c9ccafc314afaa92eee97b559a17b : SUCCESS in 14m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/01192ae67dda4eeaafd119143e851121 : SUCCESS in 26m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/24fa702c40ed4c60ba7c17a5daacc913 : SUCCESS in 15m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8ff7bac8d43b4d26a05baaf5683a1358 : SUCCESS in 15m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d2d4508e4f7b4dafa1b44df9f893d4d4 : SUCCESS in 6m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/796cd61e576d44678c3b712d4c55bc77 : SUCCESS in 8m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/509f026e97f04b8c931705bdd524a799 : SUCCESS in 10m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac2535268a314f0e9967d2c357929d53 : SUCCESS in 57m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc03d5638cb648f6a6bb1eab02230a4e : POST_FAILURE in 1h 11m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72d1520abff2492794564b263dbfc7c1 : SUCCESS in 1h 26m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/59bfb1fe9a8649a99ceb0ef8259251bc : SUCCESS in 1h 03m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8e7765f2244e3b4fd8d7a979e95c54b6a0bce088","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-30 15:20:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"274b4b8ce6eecece758e30f05d454e6b8e1817fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 16:37:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1222e8ee398d465b9e7ebfad6222e23d : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afe1cac6597e46c5bd6fb265e25ed482 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed2f05c4d28a47ab8b6c19d069b4c32b : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ed51db0bd6f4d16bbe153d03a2807b9 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b11cc76be3b547a797cf73f73d34edf0 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dac1a6c06955450d9ebb9d0ae399a064 : SUCCESS in 3m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4060eeaa12f94730a438cb1e9ba9592e : SUCCESS in 8m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e662c71a6d7f4f0385f0ca4e27586892 : SUCCESS in 9m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e2dbb47d5dde470c99d0305ab97b0cc8 : SUCCESS in 9m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5ee850fb360e4400a95521f5ed73da0c : SUCCESS in 4m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2747565c1c784a5dbf82ae3373b37bc1 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87939aa1021d4ffdb2d3f090ab3d8172 : SUCCESS in 6m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8a230cd1995d4ef3a8788dd2c284b85f : SUCCESS in 13m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1f23fa7149804457b4675fcf2ffc721a : SUCCESS in 25m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cbeaf97398ef4341a4b078670b7f42cb : SUCCESS in 12m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3428f80bc414e77a021d25311349493 : SUCCESS in 17m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3cb59f3e860049a18bad6ca5a65a5d2f : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/714a44c6bf454b5ca4b2737518609d69 : SUCCESS in 6m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5f4d31f9fbb64cc6b179a41d0bc3d4b8 : SUCCESS in 9m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40d51e6b6d1e4ef086abc66c22173ebb : FAILURE in 57m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/55e12fc14b89420baad5e0690e5cc96a : SUCCESS in 1h 03m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8b08a6f0f6a4a118d0e8dc21810de1b : SUCCESS in 1h 15m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/28cf78276d294a649ecbcfea3d58ea58 : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":1},{"id":"35feb094f2323487c72872095855f5b4eb1d6e73","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-30 16:43:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ae0e22a797e9ce02f6f137cd4d0041864eedbec6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 18:03:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2785202c0504271ba7b31d8e1b0e46e : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fde6b7f378f467aa1e7ea09f46e2647 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cdcee9a2fe2467cb4da5cad7d01096b : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b42ec0678604fa193d2dafc85164026 : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d3a9a096a2046c8b9f2ff7687042502 : SUCCESS in 5m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/089e281b5d9243959722f72adb487e93 : SUCCESS in 5m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a5daaf91166e4d4b87388fc87e1fe929 : SUCCESS in 7m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2e6c0221c770430d92ae1c768a8d0e50 : SUCCESS in 9m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e4a8c1d8c2a64d04b3b6de90cc0535cb : SUCCESS in 13m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7ce848781f3947bbaa75690eb9790d4a : SUCCESS in 4m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/45ac842361ed4e3abbe4a83a3b9e9af3 : SUCCESS in 6m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d2ed88a27b984f6db43329fab172da1a : SUCCESS in 11m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c55179ad3d0f4ab8adb83d38373fe80a : SUCCESS in 16m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/63ac3d0a1d534aef891b6fdfefbcbf5b : SUCCESS in 24m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8bb12796e69348bc849bfa3cc2624a56 : SUCCESS in 15m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a273fef1f7054ef3b4a28a31735d3345 : SUCCESS in 18m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d3db6feaaf2c4dedba73bfaffc2801bd : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/809d9f15250f423b8c07a9f57b5be446 : SUCCESS in 6m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4c8813b9fee844eba8697678a1caa42e : SUCCESS in 17m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47155201abf1472a92a07742011a44f3 : SUCCESS in 48m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9fbafd44276c4c20903fd92e1a058aec : SUCCESS in 1h 04m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d8941befefd467198d896404d38f948 : SUCCESS in 1h 18m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/3415b8a76ea5467caf58fd7675da1eef : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":1},{"id":"214a72d661dc98e46710edd9162c0bb5c1f0afd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 18:03:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c13a9b6c9ce8ba617a7ca79b44ec9ea979bfee69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 19:17:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1455588c9074a1aad8f34a59fe5efa7 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f12c59d87a74c458ff2077e3dcf20c2 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7abf8d1890340c49f402b34bb435745 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac19cd9246ce40038c6f630a168f6f68 : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1989c94805eb42c88ae86aa2d5f1ab67 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e607d917fbdd450db35852e0e161fae9 : SUCCESS in 3m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b91f20a0fc3848d3b4e6418a74a2dedd : SUCCESS in 10m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f8f9cadb97a04d66aa672ae2f80e2a37 : SUCCESS in 10m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/acf0c65251b14d86b9ab49b51e034b83 : SUCCESS in 8m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ab238773516a454a942c7fa50421cbf9 : SUCCESS in 7m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6e425a1e347249f8bbdb7dabcd0067e4 : SUCCESS in 12m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/43bfb61c6b8e489daa8e44a3a4395efb : SUCCESS in 12m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a5a25d3edbef4aee9e51fdea32d961ef : SUCCESS in 19m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a4e30b6534634b2a96ecac023d941fc1 : SUCCESS in 30m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/00d5ff687f18473292d6a5ff79023c93 : SUCCESS in 23m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d24028310e1c486fb3edc46f58cb3fc1 : SUCCESS in 25m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a573ccbaa7dd4e92ae56fc390cd38688 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e4e8f8f20ac24fd99ac46193d262f082 : SUCCESS in 6m 17s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/48829c72ccab46038db31c3b2ac0de44 : SUCCESS in 9m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/240bed56bb844b2c88fff7e8764df6af : SUCCESS in 10m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1880f0dfeb484df0be37299b80c88546 : SUCCESS in 43m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8cb753aedc884131bdbb17f140f0c547 : SUCCESS in 54m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4bef61f0fe44b39b1671397af0e0485 : SUCCESS in 1h 12m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/dd037787492541e9ad47899b04b59533 : SUCCESS in 41m 47s","accounts_in_message":[],"_revision_number":1},{"id":"69dfb0ae48a9c56eb3710223d0c2bf5b37b2d9e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 19:17:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9813bff0ffb9e791ba2310f71bfaff64e0ea6e97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 19:18:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/083128584b4744cd96601be3998a1743 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a6176bae8b8948b64c074b45c5066943c7b5da3","revisions":{"6a6176bae8b8948b64c074b45c5066943c7b5da3":{"kind":"REWORK","_number":1,"created":"2020-05-28 12:41:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/731515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/731515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/731515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/731515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/731515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/731515/1"}}},"commit":{"parents":[{"commit":"d7a824af66acbd7f540ad35e2269497eb31f9a6c","subject":"Merge \"update constraint for ovsdbapp to new release 1.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d7a824af66acbd7f540ad35e2269497eb31f9a6c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-28 12:41:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-28 12:41:12.000000000","tz":0},"subject":"update constraint for python-magnumclient to new release 3.1.0","message":"update constraint for python-magnumclient to new release 3.1.0\n\nChange-Id: I06f8ee408db610f4bd74c047de2f14b85b239362\nmeta:version: 3.1.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Feilong Wang \u003cflwang@catalyst.net.nz\u003e\nmeta:release:Commit: Feilong Wang \u003cflwang@catalyst.net.nz\u003e\nmeta:release:Change-Id: Ie55802a5a9d888ab1fb6fcba385a041e3626295d\nmeta:release:Code-Review+1: Bharat Kunwar \u003cbharat@stackhpc.com\u003e\nmeta:release:Code-Review+1: Spyros Trigazis \u003cstrigazi@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a6176bae8b8948b64c074b45c5066943c7b5da3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a6176bae8b8948b64c074b45c5066943c7b5da3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
