)]}'
{"id":"openstack%2Frequirements~731637","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I83b12d1b3b06166b2c09ce348b6daab7fbf58878","project":"openstack/requirements","branch":"stable/queens","hashtags":[],"change_id":"I83b12d1b3b06166b2c09ce348b6daab7fbf58878","subject":"Remove wheel build jobs from stable/queens","status":"ABANDONED","created":"2020-05-29 01:28:05.000000000","updated":"2020-06-01 04:01:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fa8a4b9ae30ddcfe186062c148143e4710bd4651","_number":731637,"virtual_id_number":731637,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-29 02:41:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"recommended":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-05-29 15:31:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-05-29 05:10:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-29 02:41:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-29 05:10:28.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-29 15:31:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4afb8a19f8c70a0e5d1f11395e358461ecc12f37","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-29 01:28:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1a1309c630dbe94d141412950892a1502a14771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 02:41:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/662ef4f0725d4760a36ae3df4289c18e : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/20279fa34f074631a82ae1bf26c0bef7 : SUCCESS in 3m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e94b4f80076f4be5b39c52fecccbc662 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19220dd4ddfc4c928cfe8d7843b0e8f3 : SUCCESS in 3m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/511d78886ca348348001f263431de20f : SUCCESS in 1h 12m 08s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/e6639ea333f04cbaac5b8ced06d5e823 : SUCCESS in 1h 12m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f6ea665173014bcb9a792c52843235e1 : SUCCESS in 46m 08s","accounts_in_message":[],"_revision_number":1},{"id":"31f785a48ba74e308cb6e832c0cd913fd98e3f10","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-29 05:10:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d18f5d5799fce36307d1911253c27bcbcb576521","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-29 15:31:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fa8a4b9ae30ddcfe186062c148143e4710bd4651","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-06-01 04:01:29.000000000","message":"Abandoned\n\nnow I think about it, it probably doesn\u0027t hurt to run these jobs as a test if anything happens to get updated.  we can just change the names","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3536be99bf22b76c7ff4f746c19800d7f489c727","revisions":{"3536be99bf22b76c7ff4f746c19800d7f489c727":{"kind":"REWORK","_number":1,"created":"2020-05-29 01:28:05.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/37/731637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/731637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/731637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/731637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/731637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/731637/1"}}},"commit":{"parents":[{"commit":"64f9744f3fe309a327d6c730b07a770a75f175a3","subject":"Merge \"Stop testing with pypy\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64f9744f3fe309a327d6c730b07a770a75f175a3"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-29 01:26:16.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-29 01:26:16.000000000","tz":600},"subject":"Remove wheel build jobs from stable/queens","message":"Remove wheel build jobs from stable/queens\n\nThese jobs only need to run on master; the master build job loops\nthrough the old branches from the git checkout and builds wheels.\nRemove them from stable/queens.\n\nChange-Id: I83b12d1b3b06166b2c09ce348b6daab7fbf58878\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3536be99bf22b76c7ff4f746c19800d7f489c727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3536be99bf22b76c7ff4f746c19800d7f489c727"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
