)]}'
{"id":"openstack%2Frequirements~731933","triplet_id":"openstack%2Frequirements~master~Iace8af6a397af1ae17e107fa34725fa7c70773f1","project":"openstack/requirements","branch":"master","topic":"congress","hashtags":[],"change_id":"Iace8af6a397af1ae17e107fa34725fa7c70773f1","subject":"Remove congress","status":"MERGED","created":"2020-05-30 08:12:08.000000000","updated":"2020-05-30 17:04:19.000000000","submitted":"2020-05-30 17:03:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731933-1590858181639-06738b85","meta_rev_id":"5fce9cbc48f684cb7e2c2b3c9bdbce9fbb5fb22f","_number":731933,"virtual_id_number":731933,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-30 17:03:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-30 11:41:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-30 15:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-30 15:17:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-30 11:41:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-30 15:17:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-05-30 17:03:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6dbeca5c690fbe39235f4cc74ef1f182b434e15c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-30 08:12:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2931b6a1dab6049837ca560af92e8201fd5b61e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 09:23:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c635638d69dc490b8f674d2180dce90d : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cfad3e7a11c4415a402a3ff07e7a289 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29ca5eb72d0744409f00c09a72b78655 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9759e6b05a824c13b2ec6d627cdb04fc : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc4f6e8278114db3b6d3115837ce9e3f : SUCCESS in 5m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f8f61fe0082b4ee6853245bd8bbede6e : SUCCESS in 4m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bd9d7a1ceb4c4145ba6e41a46c86c0ab : SUCCESS in 8m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4f00bd908f444a24ac68f5edb424e860 : SUCCESS in 10m 03s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/d9ccf78bd4ea4943a35c0f49e41fc990 : FAILURE in 5m 23s (non-voting)\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5a10b0cebf7946498b426738bdcd5b9d : SUCCESS in 10m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/14f0947127384769bb5df817a9d6220a : SUCCESS in 7m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/caf0c7589f704f72a0ee06f4a9fa8815 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/08b2bb957a714d53be6d5aee528b788f : SUCCESS in 9m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a96d9f022d03424db9eb2a56fdba1557 : SUCCESS in 18m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b414718e75ba464eb84083d7207d4e63 : SUCCESS in 25m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9c51ca434ccc4b0c8e9786aed6786c1e : SUCCESS in 14m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ed9d54ad574247d3a0824e7f2c8e3129 : SUCCESS in 18m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f812ced585214e63a51b6ca22d9f542c : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/931af16adb3d471b88665c1471519ad5 : SUCCESS in 8m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/726618c12b81459c9bff41511579ab07 : SUCCESS in 18m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4ffc2d2b997442ea8825cf9f1656fa14 : SUCCESS in 46m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c4044bd24754f91b3e3498503d55184 : SUCCESS in 1h 00m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5396a4a039d64ca2b04dacc525bddaa5 : SUCCESS in 1h 09m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/e8d7447c2d3b42989a866c2b9d6e821c : SUCCESS in 55m 21s","accounts_in_message":[],"_revision_number":1},{"id":"75e70a4bd53b72b36d1700dbea7d377041554866","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-30 11:41:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c4dac5675b5f68a691301c26e93a4ef9a0d4fe5e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-05-30 15:17:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"39891fb862dc4e8a62e87d132d7be48d1da68777","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 15:17:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5847dbba51f9bb1556c9bba8913de6d0eca75b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 17:03:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1c6e294d9ebd3f9b28ba3780edd87133d1368a7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 17:03:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80eb9f9e20954f85b9eedc66a2a488ad : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/636de6ffadf4491ba5264d0bf28b4540 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1420d3eadede459fb0bd9e65a88da814 : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/909a62eec97a4d2bbd9c93f38d2f7f78 : SUCCESS in 8m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5dbb0ea276134d9295848a032cc5ef7c : SUCCESS in 7m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/99684bb41edf4f9b8190ec7c749ff8a5 : SUCCESS in 7m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2a31ee15772942da9af1fe293c2c498f : SUCCESS in 15m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/00c4bc9f70c743269da3e054d27a6c3c : SUCCESS in 9m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/91da0a589d1840d59990158686abf316 : SUCCESS in 9m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8660614bd3d244c6becf50a153d4f129 : SUCCESS in 9m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/577a817769be4602bbac25fe0abf6615 : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3d189fb3115b48f3b956c57ca881531b : SUCCESS in 12m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2c66d8142726434ebcbabe8af648b3f8 : SUCCESS in 27m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2fc4c8dd23b54c93b09adb44a0d35aeb : SUCCESS in 25m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8d847d46ce484e148d9fa347f06ef0c8 : SUCCESS in 13m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/834e7eccf7c54e4581c074726e2a2e34 : SUCCESS in 23m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/254b90d3fb864b00bd08ca1f40af01c2 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/de0fc82067194ffabd70114c280df6ab : SUCCESS in 5m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/789934ed0d404555a6ccffa7c30505bd : SUCCESS in 19m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ba10c24428e6471497ee07bc50d5c58a : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/348b5748e11040568fc871c9181f91de : SUCCESS in 52m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/533db818b6074404a983f98a8b0e7ae2 : SUCCESS in 1h 05m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4233477f7f04eadb03fd7def7c8298f : SUCCESS in 1h 36m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/5acc31b2ea2a49c7a7ca0be67e758fe5 : SUCCESS in 57m 48s","accounts_in_message":[],"_revision_number":1},{"id":"de9fd506dc12a4bd277388de7dcff853ebb2c969","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-30 17:04:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bb443e6b7a048078aaf338d8d1b9f0f : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4224a11f7af687ec9f0aad5f7dfa5f0b60cbd2cd","revisions":{"4224a11f7af687ec9f0aad5f7dfa5f0b60cbd2cd":{"kind":"REWORK","_number":1,"created":"2020-05-30 08:12:08.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/33/731933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/731933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/731933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/731933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/731933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/731933/1"}}},"commit":{"parents":[{"commit":"0ae5380b6faeab73f075f9384b858f7cdbb3d4d6","subject":"update constraint for osc-lib to new release 2.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ae5380b6faeab73f075f9384b858f7cdbb3d4d6"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-30 08:11:32.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-30 08:11:32.000000000","tz":120},"subject":"Remove congress","message":"Remove congress\n\nCongress was retired, remove it from this repo.\n\nSee also https://review.opendev.org/721741\n\nChange-Id: Iace8af6a397af1ae17e107fa34725fa7c70773f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4224a11f7af687ec9f0aad5f7dfa5f0b60cbd2cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4224a11f7af687ec9f0aad5f7dfa5f0b60cbd2cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
