)]}'
{"id":"openstack%2Frequirements~732083","triplet_id":"openstack%2Frequirements~stable%2Fstein~I276d71ac835bbde7f53c6ae91fdb809e2e64b434","project":"openstack/requirements","branch":"stable/stein","topic":"wheel-cache-stein","hashtags":[],"change_id":"I276d71ac835bbde7f53c6ae91fdb809e2e64b434","subject":"Switch to build-wheel-cache jobs","status":"MERGED","created":"2020-06-01 04:06:06.000000000","updated":"2020-06-01 18:55:23.000000000","submitted":"2020-06-01 18:55:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732083-1591037723053-cca0a27c","meta_rev_id":"dd523da5494244ce7bbecda105fc10977193976e","_number":732083,"virtual_id_number":732083,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-01 18:55:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-01 10:36:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-01 14:13:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-01 14:13:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 10:36:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-01 14:13:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-01 18:55:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fa790ac1eb74306e24ff684e02171753e870ab6","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-06-01 04:06:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cf1b6661ffe1e514f55430207aaaee8e3d36113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 05:23:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11f5b80820be4057bc86edd046bde966 : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cec28c16a50947ad9a5944f0d21fa3b2 : SUCCESS in 4m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/866643cdab8d4992ad09acc6891de8b0 : SUCCESS in 8m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12410ae4fb964ee7aa54cf579cd59ae3 : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8219e44bc2b147c2863bd73636940790 : SUCCESS in 3m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/81025d07939449ecb9b25a6bc30c9e34 : SUCCESS in 56m 19s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d3ebf183916d4880bd9bccb88d55de71 : SUCCESS in 1h 16m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fefcdb0b9caf4a74bcd2f8877d186bd9 : SUCCESS in 51m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8fdafe5e549831bff1edd6c6c22da1eff3b3a9aa","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-01 10:36:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc086170491a49293182b84cf098311189701cdc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-01 14:13:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"48dee30ecaa34a0225d30ad2f9cc2ea76abe6930","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 14:14:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"531ee2d95bfcd6732372761e275d0b693ce70e7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 15:22:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43ee5acfacf144bf97c4e4bd75a15daf : SUCCESS in 7m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/faa6289abb5d4552a0448754105f7cc7 : SUCCESS in 8m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/954050abeace428098222f5a8933ddbf : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4ed92ac7a684482b5634351847ff2c2 : SUCCESS in 7m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7dc0bfaf92f4a0c9c2fbf924c033892 : SUCCESS in 6m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bedbac3dec27425aa1646fa11d953b4c : SUCCESS in 1h 05m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f6032edca0bd49e991817136005addad : FAILURE in 55m 04s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/56ece588919948598b5d2d8549505467 : SUCCESS in 40m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b44cfc112045a4038e7fbcb64df86d96d3681182","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-01 15:30:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dd5b59d568cee86e17274cc08f63532b6255590d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 17:17:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1a9188fd6d74eea8ef61a9c892a4f47 : SUCCESS in 4m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d4b26b81641740a1a89452fd635079bf : SUCCESS in 3m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/04153f7103654824bb1e01fe693bde18 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aaef3c2989224eaa8223b2ad063ed5de : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7eb21526274a42c3b5775e5b5e828728 : SUCCESS in 4m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/614672557a8645beac2ed9ae21e36a24 : SUCCESS in 1h 06m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0fb026d95b8b4bf6be57944b8f861d26 : SUCCESS in 1h 42m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/870d910dbd194e479d090592b2d406f1 : SUCCESS in 53m 07s","accounts_in_message":[],"_revision_number":1},{"id":"acfb99c5b2818eabef82aa3089ebfa087bad3062","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 17:18:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8b6f9253edff92b06b2cbf6b36f1151935e3b783","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:55:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b74525199e34532a751efec0ec63009 : SUCCESS in 8m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6f73be000cb54d1baeaf595da1c9dfde : SUCCESS in 7m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1e53d8a991fe46e086b23216e0047dbe : SUCCESS in 9m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a56980259974b409cb945364ff2e6fe : SUCCESS in 7m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/926432e4f177410491d3b6f8165fa1d1 : SUCCESS in 7m 22s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1579940f4f9e426887b9e115fa6bb879 : SUCCESS in 1h 04m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/109e2050b536416bb97adedb60de8f2a : SUCCESS in 1h 20m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/595bf55268ef43b8985b7feb4e088c0b : SUCCESS in 55m 43s","accounts_in_message":[],"_revision_number":1},{"id":"eaec8ebd98421aedebd435ce19076767a6fc759b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:55:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd3587596b05f7b8655ff24fbf8293b4a07acd77","revisions":{"cd3587596b05f7b8655ff24fbf8293b4a07acd77":{"kind":"REWORK","_number":1,"created":"2020-06-01 04:06:06.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/83/732083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/732083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/732083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/732083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/732083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/732083/1"}}},"commit":{"parents":[{"commit":"49bd519f20324bc569eb31b633dc9a41ae91c483","subject":"update constraint for neutron to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49bd519f20324bc569eb31b633dc9a41ae91c483"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-06-01 04:04:15.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-06-01 04:04:15.000000000","tz":600},"subject":"Switch to build-wheel-cache jobs","message":"Switch to build-wheel-cache jobs\n\nThese jobs moved into openstack-zuul-jobs (from project-config where\nthey were protected jobs) with\nIb41607c4842dccf27b711947bf356ef5d74a1059.  Update the old branch\nnames to use the new jobs, so we can delete the original jobs from\nproject-config.\n\nChange-Id: I276d71ac835bbde7f53c6ae91fdb809e2e64b434\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd3587596b05f7b8655ff24fbf8293b4a07acd77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd3587596b05f7b8655ff24fbf8293b4a07acd77"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
