)]}'
{"id":"openstack%2Frequirements~732085","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I94e910c12c34ef8fc5cc0553938bf2439fa8dec1","project":"openstack/requirements","branch":"stable/queens","topic":"wheel-cache-queens","hashtags":[],"change_id":"I94e910c12c34ef8fc5cc0553938bf2439fa8dec1","subject":"Switch to build-wheel-cache jobs","status":"MERGED","created":"2020-06-01 04:09:57.000000000","updated":"2020-06-01 16:45:22.000000000","submitted":"2020-06-01 16:45:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732085-1591029922384-ae35f60b","meta_rev_id":"9a93e3cd354bebe10de1036cf21e7405c80fc100","_number":732085,"virtual_id_number":732085,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-01 16:45:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-01 10:04:41.000000000","_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-01 10:36:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-01 14:14:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-01 14:14:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 10:36:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-01 14:14:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-01 16:45:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"559b18306fa179a886e05d9ce29c7f572ce272c7","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-06-01 04:09:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1702a4f90783cff4d05711abc9fc918b67659ea2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 05:32:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b5567a650fc4ce494b25667e5bf0880 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/690c046b9c1947bdaee95192eda03de7 : SUCCESS in 3m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/31a45d2b19eb4235ab8ac501ac4fc753 : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/02f88a28715441c4a485ff0a6a960183 : SUCCESS in 3m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bbdec47eba9c4803a4bbfc009da0ce68 : SUCCESS in 1h 04m 50s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/4944a7b7ec964035b165eed881846d0a : FAILURE in 1h 19m 31s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4a764c2653454519a4efedcb40c89316 : SUCCESS in 56m 01s","accounts_in_message":[],"_revision_number":1},{"id":"fe1511e85eca06ee07973db5d71229bcf6513b52","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-06-01 10:04:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9c27b6efe5f6c98bdc0d453b360b913b50c6b447","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-01 10:36:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1c30d3d07c4cae17c0ad8d1c5d60d2816c019d7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 11:34:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8899d3be30a74197b2ce38242a464a9f : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e14d5352cc5400f9f7f163bc104b025 : SUCCESS in 4m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e3aeee853b034d45bc702c6dd4e992a4 : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7c169af0c62490a8137018f39210f19 : SUCCESS in 4m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/54dc9515988541e0aad53ead98ff638a : SUCCESS in 53m 29s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/b21023d5027244dbbb9ea42607fcc63f : SUCCESS in 1h 27m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d22c1af0c43d4a06922bca1222bec987 : SUCCESS in 53m 23s","accounts_in_message":[],"_revision_number":1},{"id":"59815105f68e4948b531670d138755f39dcf264b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-01 14:14:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9aa6a828739f2b0783ad544f687d6cc722e941ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 14:14:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"371836a2c775495f4d562958e07b2cdfd53c7c3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 16:45:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0826dccc761e4708b8bf97ea000ffa54 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d42a0e474df24bd3a29f3a9be7371602 : SUCCESS in 5m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/958fb2a3e2ed421e9e9dcb4e90e7349f : SUCCESS in 7m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7fa2145215c49e08d5ebd6032c54cc5 : SUCCESS in 4m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/83f5ff0f4cf64278abcb6be9795ce7d7 : SUCCESS in 1h 01m 47s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/6233507f061d4c16ba66d5e6f1061279 : SUCCESS in 1h 23m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b26544c8c3a14d92aa271e837faf2583 : SUCCESS in 35m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b55a729e001c3ce13bb56ce6d6293e4274b61f16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 16:45:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1ed646acca0be33f221215c9eb5d0e531590bb7","revisions":{"d1ed646acca0be33f221215c9eb5d0e531590bb7":{"kind":"REWORK","_number":1,"created":"2020-06-01 04:09:57.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/85/732085/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/732085/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/732085/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/732085/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/732085/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/732085/1"}}},"commit":{"parents":[{"commit":"64f9744f3fe309a327d6c730b07a770a75f175a3","subject":"Merge \"Stop testing with pypy\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64f9744f3fe309a327d6c730b07a770a75f175a3"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-06-01 04:04:15.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-06-01 04:09:39.000000000","tz":600},"subject":"Switch to build-wheel-cache jobs","message":"Switch to build-wheel-cache jobs\n\nThese jobs moved into openstack-zuul-jobs (from project-config where\nthey were protected jobs) with\nIb41607c4842dccf27b711947bf356ef5d74a1059.  Update the old branch\nnames to use the new jobs, so we can delete the original jobs from\nproject-config.\n\nChange-Id: I94e910c12c34ef8fc5cc0553938bf2439fa8dec1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1ed646acca0be33f221215c9eb5d0e531590bb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d1ed646acca0be33f221215c9eb5d0e531590bb7"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
