)]}'
{"id":"openstack%2Frequirements~732086","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I66afc7fd27a9fdcbe3a7f2d6e50613aff0968be3","project":"openstack/requirements","branch":"stable/ussuri","topic":"wheel-cache-ussuri","hashtags":[],"change_id":"I66afc7fd27a9fdcbe3a7f2d6e50613aff0968be3","subject":"Switch to build-wheel-cache jobs","status":"MERGED","created":"2020-06-01 04:10:50.000000000","updated":"2020-06-01 16:45:21.000000000","submitted":"2020-06-01 16:45:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732086-1591029921383-9f7420e3","meta_rev_id":"20aa4255d2190aa60e3d1a5bff14a447404a5fe6","_number":732086,"virtual_id_number":732086,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-01 16:45:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-01 10:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-01 14:13:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-01 14:13:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 10:35:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-01 14:13:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-01 16:45:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"358370697924f4ea08af491361a6298ffc49c848","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-06-01 04:10:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4126f290ff69930c8446d376ea0203577cb21ffa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 05:43:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17009d2aa66f473cb69845bab401af6f : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e5c0a80255246b395c0a0eaa2247bbf : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d32529e7ab0944f8b2e130bcaec44126 : SUCCESS in 3m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/78e6e521c1884cbaae4cbc23828c3685 : SUCCESS in 3m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fed1041375e048f681500014d36e6091 : SUCCESS in 57m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed1c38a85f3041c09d7d77bb516a3376 : SUCCESS in 1h 13m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea4f48b525ee4db5ac119c9095fdb1bc : SUCCESS in 1h 32m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/daff1c579b5947a39c59873fe7644a87 : SUCCESS in 1h 13m 10s","accounts_in_message":[],"_revision_number":1},{"id":"2e7edc35baceb0326e6ea31f7b3420e10a77e3f5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-01 10:35:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"30edd85d78520690add7ea6f6e102a0952c090f5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-01 14:13:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0cb05ef8787a781629310ad667fa9a5a39a1d19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 14:14:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ccb18d8df9a5b94d90fd17373729320e168ddb4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 16:45:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a0a168e5ff05b35f7ea1acce5c81c6018e8d46d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 16:45:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7117477524164395aafbfafffd9c803e : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ed58665aa994497b908e628dac1aa1e : SUCCESS in 6m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c1602aad15004ad78b6b4e26d63c68ef : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d7b0720f5494ad4bbad815670c1a7ef : SUCCESS in 6m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/118b664691e94f41b76fa0598246212d : SUCCESS in 51m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba67c046a8f64624bd0fa48d0e61ce5b : SUCCESS in 1h 11m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25cfc93b2d66452e8af2dbf3b5468290 : SUCCESS in 1h 18m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/74361bea2690411590a573d3e1258682 : SUCCESS in 1h 07m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"491a6369f7f4ce7cdebc9441a87d8300c0a22b84","revisions":{"491a6369f7f4ce7cdebc9441a87d8300c0a22b84":{"kind":"REWORK","_number":1,"created":"2020-06-01 04:10:50.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/86/732086/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/86/732086/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/732086/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/732086/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/86/732086/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/86/732086/1"}}},"commit":{"parents":[{"commit":"9eb9f68b57b23e33e97a8318e0dc5d06e7dade96","subject":"update constraint for ovsdbapp to new release 1.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9eb9f68b57b23e33e97a8318e0dc5d06e7dade96"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-06-01 04:04:15.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-06-01 04:10:30.000000000","tz":600},"subject":"Switch to build-wheel-cache jobs","message":"Switch to build-wheel-cache jobs\n\nThese jobs moved into openstack-zuul-jobs (from project-config where\nthey were protected jobs) with\nIb41607c4842dccf27b711947bf356ef5d74a1059.  Update the old branch\nnames to use the new jobs, so we can delete the original jobs from\nproject-config.\n\nChange-Id: I66afc7fd27a9fdcbe3a7f2d6e50613aff0968be3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/491a6369f7f4ce7cdebc9441a87d8300c0a22b84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/491a6369f7f4ce7cdebc9441a87d8300c0a22b84"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
