)]}'
{"id":"openstack%2Frequirements~732087","triplet_id":"openstack%2Frequirements~stable%2Frocky~If14ebe62379e1d0d13f4c5cc7de09a1f88728ac8","project":"openstack/requirements","branch":"stable/rocky","topic":"wheel-cache-rocky","hashtags":[],"change_id":"If14ebe62379e1d0d13f4c5cc7de09a1f88728ac8","subject":"Switch to build-wheel-cache jobs","status":"MERGED","created":"2020-06-01 04:12:12.000000000","updated":"2020-06-01 16:50:05.000000000","submitted":"2020-06-01 16:50:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732087-1591030205496-1bc367d6","meta_rev_id":"d982e162e1d0ce3f1c9660f0923a0624a39687da","_number":732087,"virtual_id_number":732087,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-01 16:50:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-01 10:35:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-01 14:14:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-01 14:14:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 10:35:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-01 14:14:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-01 16:50:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"673d0b0150abbd55f899c9d369d862dd830a5c6a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-06-01 04:12:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cebf3a50d3807efdbe293b1c53e0391e4b3e84ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 05:48:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77fa5e2ca85c4c079c33ad4eae0a6150 : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f127a1f95d3a484dbff8c293b1316158 : SUCCESS in 3m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/023b0a1f2f8d410a9c62b1af04596731 : SUCCESS in 4m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/adae6388901e485693286200c6c527d8 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b65865be15a643179307b3456212c420 : SUCCESS in 4m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/be9711aee98f4a0286583a5bd3ccd5bd : SUCCESS in 1h 07m 01s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/77ff08a2664d415c8decb16112fcb888 : SUCCESS in 1h 33m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b2e8ebb82ccc4c028a7d5329a897700b : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":1},{"id":"5837278a206ee4c02b2a9b0bbfd4a40eb2051d33","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-01 10:35:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7ee7b6d9071770717a851732b39bb9c20ccc66d9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-01 14:14:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"000de8e07e533b2455a4c6c436830e9fdc0026a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 14:15:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"03d680c17f7e681cb1668d311419dad832a34f12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 16:50:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5db78b7bbf3a1142cd89f9941a892486a32b03cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 16:50:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ec0230a61f14b73896428e4b0188c47 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cd97a5f6e2f146b8bd2c7a27262257b1 : SUCCESS in 5m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/947ee35da8f9408381cd33791a29a2d6 : SUCCESS in 5m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b08b854b2f9a4a7c965f3ad6c256fef9 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ea1a6713b714469b9d6864aecc0788e : SUCCESS in 4m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b8faf2770584f9086744a0fecb60f24 : SUCCESS in 1h 05m 30s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/86b2caeab1e94600a2dbc95ef412f1aa : SUCCESS in 1h 36m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5b24e093b5454c7c995470b4d80b69ee : SUCCESS in 56m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2eea81ab8d1f731cda05cf5a547ff90b3686655","revisions":{"a2eea81ab8d1f731cda05cf5a547ff90b3686655":{"kind":"REWORK","_number":1,"created":"2020-06-01 04:12:12.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/87/732087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/87/732087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/732087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/732087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/87/732087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/87/732087/1"}}},"commit":{"parents":[{"commit":"0d587d5c1539677dc81abfc50219d602a8276226","subject":"Merge \"Stop testing with pypy\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0d587d5c1539677dc81abfc50219d602a8276226"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-06-01 04:04:15.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-06-01 04:11:51.000000000","tz":600},"subject":"Switch to build-wheel-cache jobs","message":"Switch to build-wheel-cache jobs\n\nThese jobs moved into openstack-zuul-jobs (from project-config where\nthey were protected jobs) with\nIb41607c4842dccf27b711947bf356ef5d74a1059.  Update the old branch\nnames to use the new jobs, so we can delete the original jobs from\nproject-config.\n\nChange-Id: If14ebe62379e1d0d13f4c5cc7de09a1f88728ac8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2eea81ab8d1f731cda05cf5a547ff90b3686655"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2eea81ab8d1f731cda05cf5a547ff90b3686655"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
