)]}'
{"id":"openstack%2Frequirements~732129","triplet_id":"openstack%2Frequirements~master~I53d17c4fabbf83194013297555f1f9154134b2de","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I53d17c4fabbf83194013297555f1f9154134b2de","subject":"update constraint for oslo.config to new release 8.1.0","status":"MERGED","created":"2020-06-01 09:50:25.000000000","updated":"2020-06-01 18:57:52.000000000","submitted":"2020-06-01 18:55:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732129-1591037706781-c4c60c16","meta_rev_id":"f8e8ac5d03b106d1ba41992ee652d7c6387e9b4b","_number":732129,"virtual_id_number":732129,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-01 18:55:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-01 10:30:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-01 14:14:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-01 14:14:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 10:30:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-01 14:14:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-01 18:55:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0412bbb193114ac012724d22c57412505f4e5dff","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-01 09:50:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75224a66815aca718f9f0677eb7cc81c3634c048","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-01 10:30:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"895b0d9daa8b6530a66bb15acc55dfb5d63d7ecd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 11:21:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28a4a45f10bd4fabb95ceae4ff72a0ec : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6523ce1a7cf049d982ed415aba0e9762 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/667167eb1c354b1aa13970086b5b0296 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c597631014f9408cb4f60ffe598a79b7 : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ddb0d49b74d49b0a211ee0246ce6986 : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f9049da631384debb77be9ddb7b4c762 : SUCCESS in 5m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2d07942b2d644ac19eaf61813fcc918d : SUCCESS in 8m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ae93a2618c604b9f918150a2d07192dc : SUCCESS in 9m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/10992d265a4b498d8f8d24eb15c19f2e : SUCCESS in 10m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b350853a94f04371ba258e7b681feab2 : SUCCESS in 5m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2fa0c84d308446dbbfb321dc8ec8ef02 : SUCCESS in 11m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b748c5da42a145189b2e610328c7f6e0 : SUCCESS in 8m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/04dd87279aea403f87bd5a1154b91820 : SUCCESS in 21m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5daafa2e76104cba897641371091eb5b : SUCCESS in 37m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7148a2382e0f44f6988a94d78bd79270 : SUCCESS in 13m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/767b6db2ef6d404e8ca6f0d664e5b7f5 : SUCCESS in 19m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/00e2ce7a89824be3bf1b30bde2d971ae : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b1b835f36c7a48ff82535bd252f166d4 : SUCCESS in 7m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7b0a494de1304418ba5e7f2e602521f5 : SUCCESS in 13m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30c7c786fd4547e195d4dd8baabbb554 : SUCCESS in 54m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/06f9b4270610416dbc89f29def50a491 : SUCCESS in 1h 02m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6822d6a4eb6149e1ae5044bebe5db8ca : SUCCESS in 1h 25m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/4c448b430bc14b2985bbe3ca0ddf72d2 : SUCCESS in 1h 01m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d0c002c0a4c0b22950e61d8cc07d55dfb53812ca","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-01 14:14:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"539bc315586a58a0b82188ca4c3a74dc8bedfe8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 14:15:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"175c680c024406bd733686fff4cdc2b876b55fbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:55:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdaf480d00ee418c99a579e659282aa3 : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e16d8591dc44881a2ab6a789b940267 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a755627bf0f4f929e37ae6f74927a9b : SUCCESS in 9m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a5438748697441a9a05756200c55763 : SUCCESS in 10m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a5cf4781cf0c422faa6aa77c59597bbe : SUCCESS in 7m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/837468a62f0d40f1ab390e5cd9bdccbb : SUCCESS in 6m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d2dd294daf864d208722113dec57bec0 : SUCCESS in 11m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9feb6d342f204a57bad5633733eae4ce : SUCCESS in 11m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/98a9bd7a09684d81847d1694cac9d147 : SUCCESS in 12m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/44864b7e8dd94165b96e771ccf3518ca : SUCCESS in 6m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f2ed4f33690440f49d9f80dcf28d6bc1 : SUCCESS in 10m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a2d5d35506e8494d9806451093329a23 : SUCCESS in 10m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/17d30743a3824ca689c80a1add9f39d0 : SUCCESS in 14m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5522c605aba94352bfcc89f9a095a84e : SUCCESS in 30m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0b3c21d2d145411aaa9e53a97fb4ac12 : SUCCESS in 14m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/76f20a1adbd44a57aca62b2367f584c7 : SUCCESS in 17m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e4b8ac7efcc74518a3ade3a6b5457629 : SUCCESS in 7m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/02228b4b58d54147aa4bc48065fe47e7 : SUCCESS in 7m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c22491ee66864e44a19b8f5dc07c6e76 : SUCCESS in 11m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a8b75415851148428d25ba98e465bb3e : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36f7230025e946fb8089723b81ce5708 : SUCCESS in 55m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/180587120a984a68bdeca2609863d3dc : SUCCESS in 1h 02m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab9002fbda874bf59ef6b5fcc6061b67 : SUCCESS in 1h 34m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/99aba58d80ea415db6fc967003790c2d : SUCCESS in 1h 04m 38s","accounts_in_message":[],"_revision_number":1},{"id":"52d8c14b90b7fadb022c4929bda8e416bca59839","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:55:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c9bac891e706cf0615954138ac83441674c4fc05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:57:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ab1b651b64940408e5918d8f04e176b : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd322648eda2392e6cbd0ff56b998017fc952574","revisions":{"cd322648eda2392e6cbd0ff56b998017fc952574":{"kind":"REWORK","_number":1,"created":"2020-06-01 09:50:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/732129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/732129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/732129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/732129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/732129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/732129/1"}}},"commit":{"parents":[{"commit":"df9739531bb17fbd692c355cab323db7210c9290","subject":"Merge \"Update comment in bindep.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df9739531bb17fbd692c355cab323db7210c9290"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-01 09:50:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-01 09:50:22.000000000","tz":0},"subject":"update constraint for oslo.config to new release 8.1.0","message":"update constraint for oslo.config to new release 8.1.0\n\nChange-Id: I53d17c4fabbf83194013297555f1f9154134b2de\nmeta:version: 8.1.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: Id7d99b76d4f6a95bb879c317481478f9bbbfc80a\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd322648eda2392e6cbd0ff56b998017fc952574"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd322648eda2392e6cbd0ff56b998017fc952574"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
