)]}'
{"id":"openstack%2Frequirements~732134","triplet_id":"openstack%2Frequirements~master~I7452c2b8feec6e885c58d0258f131b25e08eac17","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7452c2b8feec6e885c58d0258f131b25e08eac17","subject":"update constraint for tooz to new release 2.4.0","status":"MERGED","created":"2020-06-01 09:53:52.000000000","updated":"2020-06-01 18:32:08.000000000","submitted":"2020-06-01 18:30:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732134-1591036240943-c0d76a93","meta_rev_id":"06a526ddbaac174a61686247e383274099ea20ef","_number":732134,"virtual_id_number":732134,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-01 18:30:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-01 10:29:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-01 14:14:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-01 14:14:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 10:29:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-01 14:14:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-01 18:30:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d09922ac0ff7f422a48c8d8024544aee0943c4f0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-01 09:53:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"664a8cf7249e1a54d84765a54fc4db830b474e17","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-01 10:29:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"33476951b4046f5b258a64014f7263a5c6815e37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 11:32:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d8683862c8a4de69a670e4ca08c91ff : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd66c92211bb43e6a6048aea80c32bcf : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a4aabdf5c63417684b0fac022356f15 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c6ee57cd6794921a073e88dceb18c04 : SUCCESS in 6m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/39993c3f0eb7493a9d5c961508c2b201 : SUCCESS in 5m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f95fcd589d174accb27f5da9f7e81547 : SUCCESS in 5m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cf144bb2f9534538bc8df0c2f5f4ae24 : SUCCESS in 8m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/da0f94d028234ec787f14a945ea88e59 : SUCCESS in 11m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/68a6c1bb05ed48b1bbfed3cd845e2939 : SUCCESS in 9m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3f28588e47984508a0135101c4beb43a : SUCCESS in 7m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e1b8273ee775462fbc2f011fe3b8d5bf : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/96113c23d5ad40d594db4782a7e216e2 : SUCCESS in 9m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/44ee2c4321af4f38b49ecdf4ba2e70d9 : SUCCESS in 18m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ce361e0e08d246a3bde274b8a0dab57e : SUCCESS in 24m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4754a85877854c2f8856d7b3564ef45a : SUCCESS in 15m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c3435f7d8b1b457188eba14cef80e1c0 : SUCCESS in 19m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5b6100c4a4ea4d03af79875b34129db1 : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/39179803286a49ad8c70c944e53da4c4 : SUCCESS in 7m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dee2772f6cb8457e96698eae0ca63713 : SUCCESS in 11m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13df7d2b4cd34478af514c34b76d4046 : SUCCESS in 52m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a42a3d2b1f9e49bcb57e31eb77767a40 : SUCCESS in 1h 01m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f600c03f41546248db2f63e3724b0df : SUCCESS in 1h 31m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/03d2bf65d7e14daf9ec8d77f92dd134f : SUCCESS in 59m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ec964af7844481111c58bb9a7d270914953f12ea","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-01 14:14:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8526d52219cb427bd49229e88675fe2e7b6636cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 14:15:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e5299089f65f19438c6b09239b82d7dea486a92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:30:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6851d2c7dee4f38a4886e1d2493fb18 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68163039a5bb4c7f96c5509382af876f : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e3359ad75754a5abf004836c0526f27 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/663c232095b64d6ea74fc92442f3de24 : SUCCESS in 8m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2630833e22114067915bd055556a248e : SUCCESS in 6m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c009d5e9c0d4331b7dc98b76099c169 : SUCCESS in 5m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/55e832d9481b468da7153006db271dc2 : SUCCESS in 9m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/48081bcb1c454dcfaf11eb1b7ce4d489 : SUCCESS in 10m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/925998617a1b4f37bcd05976f0350985 : SUCCESS in 11m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/81f9e87dab61479a8e989f8e69f06d02 : SUCCESS in 5m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9c376205dc8a431691625e440abca961 : SUCCESS in 13m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/389b55bc615c47aea4ba2ac960c61f50 : SUCCESS in 10m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e8666a7aef1a4238a6232dc7e3828dcc : SUCCESS in 14m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/50811af6ddcf47c78239f1176927ffe3 : SUCCESS in 28m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/69c08d65734e449fa59a7c870fe044c2 : SUCCESS in 10m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cb2f254bbcee4400be9f52257c0a9e3e : SUCCESS in 19m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/be03cb2120584e3cbea4d92912fff6fe : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1abd5b321e5644c59941b8cc0e864244 : SUCCESS in 7m 25s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dbcea0529b8146528f4532e5fa07b8bd : SUCCESS in 11m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d8ab525d06284daa9f6d5855819c9a3d : SUCCESS in 11m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d313c64fcf2a4845a7ee64b924adf7c1 : SUCCESS in 51m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9967d1719894a7cb17b497420b1fc02 : SUCCESS in 1h 10m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b110d46927248d89edf6c5184b9de59 : SUCCESS in 1h 21m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/97ee6eade93b422e99cc5be51959da50 : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2f0e3ff4826085db2707c10941885d51183311d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:30:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"148e47fc0ca6627addd565a1f291f65a7b67281b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:32:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51696066d3fc4312a2ab28c213618b40 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f295a86f1d7e07153202be91f3ad3c9a497debd3","revisions":{"f295a86f1d7e07153202be91f3ad3c9a497debd3":{"kind":"REWORK","_number":1,"created":"2020-06-01 09:53:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/34/732134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/732134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/732134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/732134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/732134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/732134/1"}}},"commit":{"parents":[{"commit":"df9739531bb17fbd692c355cab323db7210c9290","subject":"Merge \"Update comment in bindep.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/df9739531bb17fbd692c355cab323db7210c9290"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-01 09:53:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-01 09:53:50.000000000","tz":0},"subject":"update constraint for tooz to new release 2.4.0","message":"update constraint for tooz to new release 2.4.0\n\nChange-Id: I7452c2b8feec6e885c58d0258f131b25e08eac17\nmeta:version: 2.4.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: Id7d99b76d4f6a95bb879c317481478f9bbbfc80a\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f295a86f1d7e07153202be91f3ad3c9a497debd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f295a86f1d7e07153202be91f3ad3c9a497debd3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
