)]}'
{"id":"openstack%2Frequirements~732138","triplet_id":"openstack%2Frequirements~stable%2Fstein~I621532b1b3bd92bcf96fb367929a8905f3bae2c2","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I621532b1b3bd92bcf96fb367929a8905f3bae2c2","subject":"update constraint for tooz to new release 1.64.3","status":"MERGED","created":"2020-06-01 09:57:38.000000000","updated":"2020-06-01 16:50:06.000000000","submitted":"2020-06-01 16:50:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732138-1591030206499-3595709e","meta_rev_id":"6dd58a899da2c0de24d36b4a35ed47964b8844cf","_number":732138,"virtual_id_number":732138,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-01 16:50:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-01 10:28:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-01 14:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-01 14:14:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 10:28:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-01 14:14:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-01 16:50:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fd85bd8897e5172508e1c53ed0ec309dc26112a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-01 09:57:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bab900a15c37ecf8b7fe86444d727a1b47ab25ca","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-01 10:28:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d46c6e7e308fd81a89d84d432f26d28b06b13a08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 11:11:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c910b59da88047ec9b4e1ae7deac6b3d : SUCCESS in 7m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a97f7c93914a4bb39b32f4a4b2e0e92e : SUCCESS in 6m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e4dc714793584446ac1f15807ebc3670 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56fa5b23a6f248369a43349af22ddb3f : SUCCESS in 5m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d36a8826576c40fc88273a4c64611f8b : SUCCESS in 7m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2f93549bf424414281e710f77d245361 : SUCCESS in 6m 50s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/456fa87f5baa4fdf8a558fd015a24fc8 : SUCCESS in 8m 41s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/7cd6e0c6a79249f0be554aeb0c93dd81 : SUCCESS in 9m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bac29ecb40524049b6a31692cd6bcfe5 : SUCCESS in 10m 30s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e42e4e8753f547398eb4b59c9575d64e : SUCCESS in 9m 16s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/cc8a8192c56d41f487e5d9a15641259d : SUCCESS in 10m 49s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7d629b1db720453485995f43f39cec2f : SUCCESS in 5m 44s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/2428113b844740bd9c5fd7a161499e60 : SUCCESS in 6m 16s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/baf0bd930d62458cad7d50762f9d157d : SUCCESS in 13m 06s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/475076cbe109469998f794bfdf6d537a : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fec7f4627c0343fda628f358e9a4d315 : SUCCESS in 8m 33s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4469b7bc02a34ff99ea41e565e7143a6 : SUCCESS in 25m 23s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/b7e8173e48e145e4b72ad659a69a8e12 : SUCCESS in 36m 19s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/2366aacfad234da78751da6570a280ea : SUCCESS in 28m 32s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/4f1aa419daaf4fcbbc6136c3038fe79a : SUCCESS in 30m 10s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/fab1e41fe8b9444099eea6d3e5b22ce7 : SUCCESS in 15m 22s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/06a9f388525a4dab8460b8c8be0f5f2f : SUCCESS in 16m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/13548fca2c8340f49eabfdfbda77c12c : SUCCESS in 19m 06s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/3280825c4087464891c88f5a92edc1fa : SUCCESS in 6m 46s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dfe242e185a44636b9248f6a9648cff5 : SUCCESS in 10m 51s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/b217412019b0404eabe7efe45a24e7e4 : SUCCESS in 13m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c511d64a703946029c874ca5fa4ef4dc : SUCCESS in 1h 06m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4d1841bbc54245e58467ae00214ce613 : FAILURE in 1h 11m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/513e8651c9714573a6e9163104692236 : SUCCESS in 49m 16s","accounts_in_message":[],"_revision_number":1},{"id":"7eb507ddfa11d045224de0c31f65cfb8461294d9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-01 11:34:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cb3d996150d89201afeac05fd311d0431c1a3e26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 13:03:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ef563daceed41c08dfcd936f2cb7a34 : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b61c7e5ce944f4e8e66b12721289ab3 : SUCCESS in 5m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/28ac6867844f4e438b035579656e7e74 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4566a6564c4458eaef97ab00e0e203c : SUCCESS in 6m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d7916e98a7b43a2a35a51c1838f97e5 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a5ccb5df6e04a2895cc4b0c6684b700 : SUCCESS in 4m 11s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/aa405d7bb80f42379facb2dc0ebc9759 : SUCCESS in 7m 01s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/186f45c557104d1fb6077477138488e6 : SUCCESS in 8m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/518f76a017944f46b5f22eb4cd26e4a6 : SUCCESS in 7m 44s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e5e178f030d049518454dc2a8c3f8bcc : SUCCESS in 9m 18s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/737102ea2fff4398b877efe302ee425c : SUCCESS in 11m 35s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/803b8025d6964ed585b065755a91d011 : SUCCESS in 5m 50s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/05ed54ff31f140f4b4555d124c912b72 : SUCCESS in 4m 21s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/86f10050b7d84382bcaba95a4a83ae16 : SUCCESS in 11m 08s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/7e76ea46187745069de60ed986bb1813 : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ebb484bbe59460a9d8a2f8e9063ab8e : SUCCESS in 7m 57s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c4490ecb84b34750b4cff5afade4949c : SUCCESS in 23m 24s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/34780fe0c35141f2b5ed9442242f072e : SUCCESS in 29m 37s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/44d9f7a8fbd447349b625217cd7fd915 : SUCCESS in 20m 49s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/98b76ea68f004fed81942dfa33d66c67 : SUCCESS in 22m 43s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/44a7764a942c4a54bf49770f1e91a480 : SUCCESS in 13m 58s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/2647be08dcc14134acdae329959182c0 : SUCCESS in 12m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f26332d36d044d01a1403ea42eb41f0d : SUCCESS in 22m 04s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/7018ea3fae5d4d0dbe3d4010bf2ed492 : SUCCESS in 4m 40s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/933e6872ead84380afb9666955f8c536 : SUCCESS in 8m 37s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/0896e3f96f4642da9a9565dd5a05a8d8 : SUCCESS in 14m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/de999fae970244a4b5893c0c73498110 : SUCCESS in 1h 03m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3ff620bfdf0745bb956749f70b8f6817 : SUCCESS in 1h 24m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3f4f5b5686cd4ae0bc5c0e48e77f2d9a : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7d6f189285ca1a30ae8cf18c20b1d5b16f2f35bc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-01 14:14:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"447fc2f8d66a137308658a04c3017d9fd35b6afc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 14:15:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4479ebbca418968168d1512b880ef2574397b993","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 16:50:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf0714c44d54be3b713cf62f749a399 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/75b185de2fc64117a5d71a51d1dfb5d8 : SUCCESS in 6m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6819062df36d4dfaa79d238f0282c4e2 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/291b3ed405714a209c9afea8f100e55c : SUCCESS in 7m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c01eb206e634be8a0c5e77dd5e20c28 : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/95bf371270384c088faf77c4e02f1687 : SUCCESS in 5m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e4c268a46f5f41ccac77db7b7f60a337 : SUCCESS in 10m 29s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/bf7409dff9d643b5a8dc93fb42808aa7 : SUCCESS in 9m 08s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b3054321209a4655b881413ceccda979 : SUCCESS in 12m 00s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/48f56c05de534cb4ada5230f9a18251e : SUCCESS in 11m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/89075cf96b324830b902f3dff5376971 : SUCCESS in 7m 34s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/bbe67f736b644c229a0f62ed18e2262f : SUCCESS in 6m 55s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/fa306a927c6444b9948bd9af96875832 : SUCCESS in 13m 44s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/cb95c291102d47a6b91118c134f606ff : SUCCESS in 12m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/01a377227f7e48b18de86f67127a7ca9 : SUCCESS in 8m 05s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/09b53e5ed4524af6a8391eeac439a953 : SUCCESS in 25m 00s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/13af9682bae74655a1ae5c5e4ddf86b5 : SUCCESS in 29m 17s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f46b48425d1c46df8cc8f41e201fbc72 : SUCCESS in 22m 42s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/d6d98fef27fe45328d36388e221b0290 : SUCCESS in 23m 25s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/82f69b5bbb3f4964883b7759b546d5b7 : SUCCESS in 16m 39s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/7c032f16356047b198981e72183863ef : SUCCESS in 16m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5b70443cc5604830b549488aa6bf47a4 : SUCCESS in 19m 11s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/debd6d85d19b4c468226f736ba64e4a5 : SUCCESS in 5m 45s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/97d3cf03a27c41f087f58f5a9713f7cf : SUCCESS in 9m 53s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/7d9f93fecf034d64b773605d2b19f2f9 : SUCCESS in 10m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/db7b051f921c4068923a1a80b4fbf238 : SUCCESS in 1h 15m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dc8979087b814b1cb115450c7166debf : SUCCESS in 1h 26m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c3e97f676c2241f4b236a861fa4fcc8b : SUCCESS in 1h 06m 49s","accounts_in_message":[],"_revision_number":1},{"id":"037800e41b60e4bd62d131d92b872da9e3bde848","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 16:50:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"227bd2c452ed7c5c160eedea99f5c76b955cc4bc","revisions":{"227bd2c452ed7c5c160eedea99f5c76b955cc4bc":{"kind":"REWORK","_number":1,"created":"2020-06-01 09:57:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/38/732138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/732138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/732138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/732138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/732138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/732138/1"}}},"commit":{"parents":[{"commit":"49bd519f20324bc569eb31b633dc9a41ae91c483","subject":"update constraint for neutron to new release 14.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49bd519f20324bc569eb31b633dc9a41ae91c483"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-01 09:57:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-01 09:57:37.000000000","tz":0},"subject":"update constraint for tooz to new release 1.64.3","message":"update constraint for tooz to new release 1.64.3\n\nChange-Id: I621532b1b3bd92bcf96fb367929a8905f3bae2c2\nmeta:version: 1.64.3\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Change-Id: Ib100112c2882f3d2c3be1247b2df5474b5f7bbac\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/227bd2c452ed7c5c160eedea99f5c76b955cc4bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/227bd2c452ed7c5c160eedea99f5c76b955cc4bc"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
