)]}'
{"id":"openstack%2Frequirements~732516","triplet_id":"openstack%2Frequirements~master~I0c339b4202299521468ef657a3a3de9a68b9cd09","project":"openstack/requirements","branch":"master","topic":"add-daemonize","hashtags":[],"change_id":"I0c339b4202299521468ef657a3a3de9a68b9cd09","subject":"Add daemonize library","status":"ABANDONED","created":"2020-06-02 10:11:14.000000000","updated":"2020-06-09 08:13:24.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"003ae03fdb1f67e984b9bc1215225349c21deca7","_number":732516,"virtual_id_number":732516,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"date":"2020-06-09 08:13:20.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-04 15:26:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2020-06-04 19:05:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 12:14:31.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-03 08:16:03.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-04 15:26:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 19:05:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"053b4a57904bc09ff01a19ae020b0b20079683d4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-02 10:11:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9b5948457375db9fed1fc3360ed0f3fbf114bae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 11:31:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7488626f55b43919c4f67db7d698407 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5d49e3c12ae452199a5b870b5e5f0ba : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e1b8ae22a704ca6bc5f5393fe2e70a3 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb0e3624b4334408a889c1b560aeb1f2 : SUCCESS in 7m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/990856a96f8d48498050aa73fb87398c : FAILURE in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7af9a7caf8344f339cb5b463e377d89c : SUCCESS in 6m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3944068148184a40a7a55aedb32c0c32 : SUCCESS in 8m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d773cd77025c4be38df97a65d64df05d : SUCCESS in 9m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3bd1cc9dee5a4d91bff9ab360144e903 : SUCCESS in 10m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/470b3449baa743f09f3ccbc2876ef7a1 : SUCCESS in 7m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/75026f5a04d34baf801bc0fba80edaaa : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95dcb1ccbf084e5b99979af7de4917b8 : SUCCESS in 8m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4fe8d75715b1404da64f6c9494f70fe3 : SUCCESS in 27m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8fb0bf210ba7426e965b49934403411b : SUCCESS in 26m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eb85e91cc5bb483198300a83e2190ddd : SUCCESS in 22m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f1b135b1c7e94c448318d2af148cdde9 : SUCCESS in 17m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/800f167c39ee4205a8f7284962bf9d70 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c6da16a40a4410491fc50cdd1c0df4d : SUCCESS in 6m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e87544e3de8e42f993de5c771d8f410a : SUCCESS in 9m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82dae21f303e442e9239c73c5fa822b9 : SUCCESS in 54m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f62142e4305847eb8a66e881751a00a5 : SUCCESS in 56m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35688f1b3d4d40bc99e65cf2aa7601a5 : SUCCESS in 1h 14m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/880591f0cb0d4eb081b30143a404a467 : SUCCESS in 1h 07m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b167f001bae92ec2614592498675fda7c2be3cfd","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-02 11:50:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2f8abf1c9417c3b5e10857ccf31161c57beff99d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-02 12:14:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c266af08f304165658035d86fa1cbd6e9bc62cfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 13:11:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8fb3f266f53405780f0a8b00bd053cb : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2d86b0800b146cc9359e4273f2baf5b : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2729f0ed279544efadb073917b5c4201 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a112f32f07f5462abaa8ebec8f86772a : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f732d3d773743c686ae17a2bd657e90 : SUCCESS in 3m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e120582e7c84121afd225fdedd1fcec : SUCCESS in 3m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/268865e551e74952a69548fbeff99e10 : SUCCESS in 11m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/56e699fb966549d182f91b8a90eca39e : SUCCESS in 9m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3da63bfb072b4723a2addde6072acd3b : SUCCESS in 9m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1b139c05465e452289ae1de906e1a692 : SUCCESS in 6m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/32cc17daa03c4358960377c3190247fb : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2536c6bcc424a268f324700fee57a65 : SUCCESS in 8m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e748f6fb16e04455a201071e8d6d0f95 : SUCCESS in 13m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/baf136e9f3df4953b7077b20ef1ad80e : SUCCESS in 25m 58s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8c39bc0320644f2a9a45a145b5eac533 : SUCCESS in 18m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8a3d74edda8a4415893ffcd39ed7ba33 : SUCCESS in 17m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/da899eeb1fd14a15a1b986101ec5613b : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/130681c16e684b7ca444e4db3283bb72 : SUCCESS in 6m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7c912846eb3e46cfa77d76cf1e96a0b9 : SUCCESS in 9m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d756319161dd43a4acbc03cb6eb92191 : SUCCESS in 52m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/af49b1a41a674e56ba3a172bf6374c70 : SUCCESS in 45m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3d7cf70987d4b89a091ca7be7fee7a1 : SUCCESS in 1h 15m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/953c3849952240b98f0db2f22464831f : SUCCESS in 1h 10m 42s","accounts_in_message":[],"_revision_number":2},{"id":"368231b88513e7191a4830d84702f31d89ae7cd2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-04 08:53:12.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e603d1bd4afd2eaf28f7766132226db957a4995b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-04 08:56:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cec40a456b8cbe34888b3ba52318df7190b173e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 09:57:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b44d24dc6ce4dbbb2eb0cc087d56f25 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aef8bbf22c404305826b4ec5192d525e : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41cc1d8ee33e4cb3962eb95dd5ff0914 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e672d807b55c437b9f2e6cb05a298553 : SUCCESS in 6m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f580d2d2d79b484da0206cffa5e41204 : SUCCESS in 4m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c14d62f2c5d548c78bd1bbda34f16f59 : SUCCESS in 6m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0d01f0582eea4f25bedd0c75f7dc94c2 : SUCCESS in 9m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a003f615ec3e4f08b5c4a7984c69269f : SUCCESS in 9m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7ab96680bbe14f9fb8c2a71759e2d717 : SUCCESS in 9m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9e89e0eb07ac46ddb1bb0fea5fc2fedf : SUCCESS in 5m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5746da15144146c7a9bcd38c2b73dec0 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d18bf3b8ddf54efeb13e9f2f260ce6d3 : SUCCESS in 7m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/63ab48bee9e64520b7ac67cfb594fbe6 : SUCCESS in 13m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0d389013a8564198b05a865eab21182d : SUCCESS in 26m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/023d46ac8f0440e5877f3adf4f892ff9 : SUCCESS in 16m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eb5a7305efb64f1c841b5dbbeee95bc2 : SUCCESS in 17m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0140db30eea64b16a122737778d83e33 : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a6984a76931e405eada2e8ad79e911c5 : SUCCESS in 7m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bbf67ca2f32a497188f110c7d6513620 : SUCCESS in 10m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/addba10d5e9e4ec5b0c62871f85bf33c : SUCCESS in 51m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b8c21b8bcc7f429689e15147b8e76b26 : FAILURE in 50m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc69a6ec1b384592842adfacadeec98c : SUCCESS in 50m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/e65793d57ad040d6bd1893baca2623e8 : FAILURE in 1h 02m 38s","accounts_in_message":[],"_revision_number":3},{"id":"d0cbb0556afdaab60b1d42c4b9f0bf8bcee188aa","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-04 13:01:49.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e84403f65117e97bbd4d1105a05767a3c6eb5ca8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 15:26:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1970dba5c2944242b720a80ad6c1b4c7 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b02026e7a54d4e7db91a068e7b8aecfe : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d48f4d6b57ea4a2c8ce3b788e4c3bcf3 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34478fac859b4ec981106d16f77769b3 : SUCCESS in 6m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b56615287d54f01a2d7c835422e5111 : SUCCESS in 6m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/391f2343425c437787fe4501492a806d : SUCCESS in 6m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0769017a117e4557ac212e5fb94b5897 : SUCCESS in 10m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dbc5a498f7b443208803775bc29b143b : SUCCESS in 11m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c58e4962d2224d69a0fbccb1e16b94e4 : SUCCESS in 8m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4e5a1d91d2f74e429ff71e89dba7b612 : SUCCESS in 6m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a14ff8e1d9cf46d08dbe928121ba79c5 : SUCCESS in 12m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/de7e1c2d5a054617b31474a978e8b0de : SUCCESS in 10m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e887f095d3654da789049a91211bf4e7 : SUCCESS in 12m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fba7c5cfd26f40638980091c5e838557 : SUCCESS in 28m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3f7710c2a18f44249a78270434d0ab95 : SUCCESS in 18m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3c940b911fe47658ff956c4aac2fe7f : SUCCESS in 19m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/eb83e98597414438a927fa20ceda5982 : SUCCESS in 6m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ea99f64de00742d8bfd5af21497ef797 : SUCCESS in 10m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/557ccabd70134d018f43430b6bcfc16e : SUCCESS in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/453538d1689b4a999dc64db31b8b4fd2 : SUCCESS in 58m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9f7b7f1c68b24a7f9454c61845a61fce : SUCCESS in 1h 34m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81661618e09b4f799eac147260f06f54 : SUCCESS in 58m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/3f0078783ed5495cb2c50952f74aedc4 : SUCCESS in 47m 01s","accounts_in_message":[],"_revision_number":3},{"id":"549e3634993a3a80b1f1af12d9d98dff5a3f4f52","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-04 19:05:41.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"766e29e221cb47eda2e8e75193eefce0e87c1a80","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-05 07:11:33.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"910caf2858824cf5687da4cb322cb8a5bf379364","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-05 14:53:55.000000000","message":"Patch Set 3:\n\noslo.service probably (or more directly one of it\u0027s requirements, but I\u0027d just use oslo.service for standardization reasons.","accounts_in_message":[],"_revision_number":3},{"id":"8fccf021926fc433fe54ed6482b3f69530f7dc5b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-09 08:13:20.000000000","message":"Patch Set 3:\n\n\u003e oslo.service probably (or more directly one of it\u0027s requirements,\n \u003e but I\u0027d just use oslo.service for standardization reasons.\n\nthanks! I\u0027ll check that","accounts_in_message":[],"_revision_number":3},{"id":"d48f2e7e13c79a25a717e057ce8dffb5293ceb8f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-09 08:13:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a790193016fdba1a5ae13536c6b3edc3e9fb6275","revisions":{"ecb64eaf99bb31d938452f1fac2ec235445311a3":{"kind":"REWORK","_number":1,"created":"2020-06-02 10:11:14.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/16/732516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/732516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/732516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/732516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/732516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/732516/1"}}},"commit":{"parents":[{"commit":"cc76e4f7173b2157a62163377c4d595d28309707","subject":"Merge \"update constraint for openstackdocstheme to new release 2.2.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc76e4f7173b2157a62163377c4d595d28309707"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-02 10:10:48.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-02 10:10:48.000000000","tz":120},"subject":"Add daemonize library","message":"Add daemonize library\n\nThe daemonize library is still used by molteniron\n\nChange-Id: I0c339b4202299521468ef657a3a3de9a68b9cd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecb64eaf99bb31d938452f1fac2ec235445311a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ecb64eaf99bb31d938452f1fac2ec235445311a3"}]},"branch":"refs/heads/master"},"62baa4a022cda15398b92846bc840e49cb595511":{"kind":"REWORK","_number":2,"created":"2020-06-02 11:50:42.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/16/732516/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/732516/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/732516/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/732516/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/732516/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/732516/2"}}},"commit":{"parents":[{"commit":"cc76e4f7173b2157a62163377c4d595d28309707","subject":"Merge \"update constraint for openstackdocstheme to new release 2.2.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc76e4f7173b2157a62163377c4d595d28309707"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-02 10:10:48.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-02 11:50:33.000000000","tz":120},"subject":"Add daemonize library","message":"Add daemonize library\n\nThe daemonize library is still used by molteniron\n\nChange-Id: I0c339b4202299521468ef657a3a3de9a68b9cd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62baa4a022cda15398b92846bc840e49cb595511"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62baa4a022cda15398b92846bc840e49cb595511"}]},"branch":"refs/heads/master"},"a790193016fdba1a5ae13536c6b3edc3e9fb6275":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-06-04 08:53:12.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/16/732516/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/732516/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/732516/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/732516/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/732516/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/732516/3"}}},"commit":{"parents":[{"commit":"cc76e4f7173b2157a62163377c4d595d28309707","subject":"Merge \"update constraint for openstackdocstheme to new release 2.2.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cc76e4f7173b2157a62163377c4d595d28309707"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-02 10:10:48.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-06-04 08:53:07.000000000","tz":0},"subject":"Add daemonize library","message":"Add daemonize library\n\nThe daemonize library is still used by molteniron.\n\nIt\u0027s in a good state, but not maintained since roughly 1 year.\nIt\u0027s Python 3 compatible, officially up to Python 3.5.\nIt\u0027s released under MIT License, so license compatible.\nIt\u0027s already packaged in the distros we target.\n\n\n\nChange-Id: I0c339b4202299521468ef657a3a3de9a68b9cd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a790193016fdba1a5ae13536c6b3edc3e9fb6275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a790193016fdba1a5ae13536c6b3edc3e9fb6275"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
