)]}'
{"id":"openstack%2Frequirements~732945","triplet_id":"openstack%2Frequirements~master~I15d451f6a832b2e698b28e7351a36ef7aea92abe","project":"openstack/requirements","branch":"master","topic":"drop_future_imports","hashtags":[],"change_id":"I15d451f6a832b2e698b28e7351a36ef7aea92abe","subject":"Stop to use the __future__ module.","status":"MERGED","created":"2020-06-02 18:56:13.000000000","updated":"2020-06-03 03:14:04.000000000","submitted":"2020-06-03 03:12:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732945-1591153966212-f8d2b182","meta_rev_id":"ccf66e7b30bd573322ed23a720d54de55648b5c8","_number":732945,"virtual_id_number":732945,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-03 03:12:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-02 19:02:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-03 00:56:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-03 00:56:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 19:02:27.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-03 00:56:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-03 03:12:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b11777aeee9d1b3fcf56d8d9bffc62b5cf4df769","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-02 18:56:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fab22fb9780bb4129f275fef4be3c6415e4e21d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-02 19:02:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f660509a99b7a321c29801f1c37fe5ce81ec3130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 20:37:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a02fce6943748e3a663ed5aabbb47b1 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f1e18fbb5ef4ad0aa79fd812d1f4c15 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c37fa27677e47e796d473e28f72b9a1 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdfd08c372cc486880f640deaab2da4a : SUCCESS in 7m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ec07f58f47f4dc09753148c79abdcc3 : SUCCESS in 4m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2497f10494874eb896bb5e0ddbfdf5ae : SUCCESS in 55m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d28c40d9894248f992b892f598481d36 : SUCCESS in 1h 06m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2e23dfa97614022b8a88425a416ad66 : SUCCESS in 1h 39m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/3fff88ca9c384e3fa75c4b9bb44db0f3 : SUCCESS in 1h 11m 07s","accounts_in_message":[],"_revision_number":1},{"id":"30a1a899d43ede805f0cd47a2036a9a295797e9d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-03 00:56:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"21f0a3271e1e396a3b1e171cfffe2e3200a1989d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 00:56:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ce6c1deb33b38b094192e7ac3bcfc9b945d2bfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 03:12:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c56809c31446c1aa4b0cb2f51c219b : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3ffd61421244f128ba7506a01497f9d : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b49ed3766d3a46b8a99e9ee80923cac7 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/409c6d1a2e1945608a963bb852addf02 : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f8ed6475c96f45bea4400c6cf927c01e : SUCCESS in 4m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4657880ed3c84ba7a1bb2f0a77857faf : SUCCESS in 56m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/60094e674ff64ff899a6888350d88506 : SUCCESS in 1h 14m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/53d70f411af44c50b9f9ebc46f2f5c25 : SUCCESS in 1h 06m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/3af114fd0a8b439e94ca561ed676d812 : SUCCESS in 1h 05m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c954203dbdc67b820b2115017c45050245517761","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 03:12:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0483f65dc33f2d0daae184ea270b199305e1e506","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 03:14:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe2d1277340d46ed84cafcba2455e408 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fac394f79cbf9213a340fdc887a5ed58d5262886","revisions":{"fac394f79cbf9213a340fdc887a5ed58d5262886":{"kind":"REWORK","_number":1,"created":"2020-06-02 18:56:13.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/45/732945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/732945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/732945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/732945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/732945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/732945/1"}}},"commit":{"parents":[{"commit":"f271b7646d977d6a2b71d153637a9b208a6b1791","subject":"Merge \"u-c: update protobuf to 3.12.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f271b7646d977d6a2b71d153637a9b208a6b1791"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-02 18:56:04.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-02 18:56:04.000000000","tz":120},"subject":"Stop to use the __future__ module.","message":"Stop to use the __future__ module.\n\nThe __future__ module [1] was used in this context to ensure compatibility\nbetween python 2 and python 3.\n\nWe previously dropped the support of python 2.7 [2] and now we only support\npython 3 so we don\u0027t need to continue to use this module and the imports\nlisted below.\n\nImports commonly used and their related PEPs:\n- `division` is related to PEP 238 [3]\n- `print_function` is related to PEP 3105 [4]\n- `unicode_literals` is related to PEP 3112 [5]\n- `with_statement` is related to PEP 343 [6]\n- `absolute_import` is related to PEP 328 [7]\n\n[1] https://docs.python.org/3/library/__future__.html\n[2] https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n[3] https://www.python.org/dev/peps/pep-0238\n[4] https://www.python.org/dev/peps/pep-3105\n[5] https://www.python.org/dev/peps/pep-3112\n[6] https://www.python.org/dev/peps/pep-0343\n[7] https://www.python.org/dev/peps/pep-0328\n\nChange-Id: I15d451f6a832b2e698b28e7351a36ef7aea92abe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fac394f79cbf9213a340fdc887a5ed58d5262886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fac394f79cbf9213a340fdc887a5ed58d5262886"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
