)]}'
{"id":"openstack%2Frequirements~733008","triplet_id":"openstack%2Frequirements~stable%2Fstein~If06d9fa98975979ffeda5be4c7b6b805a7915d95","project":"openstack/requirements","branch":"stable/stein","hashtags":[],"change_id":"If06d9fa98975979ffeda5be4c7b6b805a7915d95","subject":"Pin diskimage-builder \u003c 3.0.0 on python2.7","status":"MERGED","created":"2020-06-02 21:43:48.000000000","updated":"2020-06-03 02:22:21.000000000","submitted":"2020-06-03 02:22:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733008-1591150941992-f8d2b182","meta_rev_id":"7add9bab9edf1cfd57f4fab32911b0d9d121d365","_number":733008,"virtual_id_number":733008,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-03 02:22:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-02 23:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-06-03 00:41:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-03 00:41:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 23:58:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-03 00:41:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-03 02:22:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d720f2d9485686716fade3aa5f3a67f23b4ad5da","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-02 21:43:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d957209682c12141f0c472402e25fa24d886c5dd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-02 21:54:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3ebc7f8eb77a3d26ba8a606aa5d75e51368b926f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 23:10:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77315136f14f4c1b8c2abf9e3518a652 : SUCCESS in 6m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/acd0642dd7a141009de4ffd7a625fb27 : SUCCESS in 4m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7eb4d1036d7d4893bc7292e469d8118c : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c664bdb1651e48b596af798fdd50a6dc : SUCCESS in 6m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9373165cb7e94c2c803f2e5f31d70264 : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/00c2590ae9e34553943b8fa57d0db9bd : SUCCESS in 5m 42s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c7d84af8086a4855adc969d954e2ac37 : SUCCESS in 6m 54s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/6f04561a822e450fa778bc65000423c5 : SUCCESS in 10m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3b3d8f3473a64955b8ea82486dc7503e : SUCCESS in 7m 27s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c82cf32fa6ad48c5a4834980312e7ec4 : SUCCESS in 10m 26s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/bda7fb992b0447aebeebc8f43085f9c6 : SUCCESS in 10m 34s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/085271ccf9b542f09db7cd79a44e273a : SUCCESS in 5m 31s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/2437cec9c0b244d78bb786df30b2eb99 : SUCCESS in 5m 41s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/21280d4be0124b75a495744d1a1a41ac : SUCCESS in 11m 53s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/baf07ac1ea2d4c4984e73022177d6c72 : SUCCESS in 11m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/057f3b54c2564556b1ab1397d0af5c09 : SUCCESS in 8m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/54292accedab4e9ab86f9cf164bf0f9f : SUCCESS in 21m 26s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/a6cfb29237154c12868f9766635557c9 : SUCCESS in 26m 10s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/64efe3a8ebb74c8298332f386d5f2491 : SUCCESS in 23m 55s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/5c36ebe184704ad9873874fb8eb0b014 : SUCCESS in 14m 08s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/88c139ddb11149229c9554e8f411908e : SUCCESS in 10m 00s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/2afaf5a8e88d45fca9e175660ecbf0a7 : SUCCESS in 14m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/495b5e9dc4f44b699753b06a46d1a7d0 : SUCCESS in 20m 36s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/e9873ea198604fecb4fceae1970487cc : SUCCESS in 5m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/e7ae1456d0174d8ca0b6ba44009ca646 : SUCCESS in 9m 17s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/f14acee1d0d4472fa83bdda2a75dd0ed : SUCCESS in 12m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/10dc7796ed2342acb6edfd5e93a05b6f : SUCCESS in 1h 10m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d14df7d963984cc98f644f5ff0e58432 : SUCCESS in 1h 00m 09s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0dca85baaad94845b37ccec50c4a9724 : SUCCESS in 51m 12s","accounts_in_message":[],"_revision_number":2},{"id":"6e146e369d6fd0a1083ad536e3479707399e3901","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-02 23:58:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c8858a77e34c30543eff5870fbad9c3e227b26dc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-03 00:41:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7a88460586586240de8f4fde1e60a400b1fd92a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 00:41:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1258f8492e46d96ba0712e8479272dcc7b3cd071","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 02:22:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7add9bab9edf1cfd57f4fab32911b0d9d121d365","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 02:22:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/044ca74c5cc44549b96454c4d285690a : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69a0d1948d9f470b93512238dd668807 : SUCCESS in 4m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/33076883ae0f409bbd608e9b39041cdb : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9794cd5b0e834527bc7c94eaac4c2c49 : SUCCESS in 3m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9528e5a433d445f97550ebc7e6f72d6 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fe9921fbbb904466b9b4b07784ac6c09 : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/a93a05cbd2914460b6fe9ab53ee538ee : SUCCESS in 7m 52s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/a754fe207e47406993593de3fedd6a7e : SUCCESS in 7m 25s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9da4afe1b5bf449889b3831c1c0657a9 : SUCCESS in 9m 55s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/30683f25f8764008a24b52d3c2702a9e : SUCCESS in 10m 25s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e6afd0ab952e4b4a87c3c1e7b36c2059 : SUCCESS in 4m 27s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/d0c5cc1486884304a0d2254c8d11f1d3 : SUCCESS in 5m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/bbfc02e5b49b49fea81c59f769566519 : SUCCESS in 9m 52s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/4d3e22cd07ca496f98adf67ce34b7dff : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a9fa5155beb64dfa80a0b2d0b476e51f : SUCCESS in 7m 04s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/14ae8053e42b45db9433f12d95849b10 : SUCCESS in 21m 42s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/494b05a9d7f14b2ea77b58963a3ab836 : SUCCESS in 24m 18s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/3204dd3dd7624c1e91d419c7bc2e9996 : SUCCESS in 22m 02s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/a0a94c207413472185d0ba704fde6fc6 : SUCCESS in 26m 31s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/95003abbd1004280a0787d94c6f496b5 : SUCCESS in 9m 42s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/483345c1c8234acb8ac15f46f25f14dd : SUCCESS in 13m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/201319ea8bff40119dc73d261dcbfa0b : SUCCESS in 16m 45s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/a49c91ddca02421e9f5f4b5cec31f129 : SUCCESS in 5m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5b494a0a027a4b3dbc5d8a58a51ec566 : SUCCESS in 8m 55s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/73cb505ad2cb4068a267d30907b63538 : SUCCESS in 7m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0fb83daa676b46d0ae5df1edc14c7c94 : SUCCESS in 1h 05m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8ee5d5e76dde4107a20b0d1c76faf04d : SUCCESS in 1h 40m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cc38b50528964644a8e45906741b8a68 : SUCCESS in 57m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"49449167f14400b5f79cedbb02ea69fe5a0fec29","revisions":{"501dc4fc8c590a4c2572b674c64a56419d282b49":{"kind":"REWORK","_number":1,"created":"2020-06-02 21:43:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/733008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/733008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/733008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/733008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/733008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/733008/1"}}},"commit":{"parents":[{"commit":"30c521988281f31dab3b1899a27cb0a246b1ae55","subject":"Merge \"Switch to build-wheel-cache jobs\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30c521988281f31dab3b1899a27cb0a246b1ae55"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-02 21:41:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-02 21:43:47.000000000","tz":-420},"subject":"Pin diskimage-builder \u003c 3.0.0 on python2.7","message":"Pin diskimage-builder \u003c 3.0.0 on python2.7\n\ndiskimage-builder has dropped support for python2.7[1] in version\n3.0.0. We need to pin it to \u003c 3.0.0 for the stable branches that\nstill use python2.7 and diskimage-builder.\n\n[1] https://review.opendev.org/728889\n\nChange-Id: If06d9fa98975979ffeda5be4c7b6b805a7915d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/501dc4fc8c590a4c2572b674c64a56419d282b49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/501dc4fc8c590a4c2572b674c64a56419d282b49"}]},"branch":"refs/heads/stable/stein"},"49449167f14400b5f79cedbb02ea69fe5a0fec29":{"kind":"REWORK","_number":2,"created":"2020-06-02 21:54:05.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/08/733008/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/733008/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/733008/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/733008/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/733008/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/733008/2"}}},"commit":{"parents":[{"commit":"30c521988281f31dab3b1899a27cb0a246b1ae55","subject":"Merge \"Switch to build-wheel-cache jobs\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/30c521988281f31dab3b1899a27cb0a246b1ae55"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-02 21:41:47.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-02 21:53:59.000000000","tz":-420},"subject":"Pin diskimage-builder \u003c 3.0.0 on python2.7","message":"Pin diskimage-builder \u003c 3.0.0 on python2.7\n\ndiskimage-builder has dropped support for python2.7[1] in version\n3.0.0. We need to pin it to \u003c 3.0.0 for the stable branches that\nstill use python2.7 and diskimage-builder.\n\n[1] https://review.opendev.org/728889\n\nChange-Id: If06d9fa98975979ffeda5be4c7b6b805a7915d95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49449167f14400b5f79cedbb02ea69fe5a0fec29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49449167f14400b5f79cedbb02ea69fe5a0fec29"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
