)]}'
{"id":"openstack%2Frequirements~733010","triplet_id":"openstack%2Frequirements~stable%2Frocky~I681172d66891b69b5cd538d0598968262d90e368","project":"openstack/requirements","branch":"stable/rocky","hashtags":[],"change_id":"I681172d66891b69b5cd538d0598968262d90e368","subject":"Pin diskimage-builder \u003c 3.0.0 on python2.7","status":"MERGED","created":"2020-06-02 21:46:30.000000000","updated":"2020-06-03 06:59:21.000000000","submitted":"2020-06-03 06:59:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733010-1591167561934-edd24659","meta_rev_id":"a9f580606965cae07a495fce6f919a6f4b5e936a","_number":733010,"virtual_id_number":733010,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-03 06:59:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-06-02 23:58:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-06-03 00:41:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-03 00:41:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 23:58:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-03 00:41:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-03 06:59:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d77ccd2651ca4c157d9b20868a22ddfa4aaad82","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-02 21:46:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ae61effc9fb9e31266f033b1c24fd5eccd6d008","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-06-02 21:54:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a57e931e74d4b63bc308a6d4c1fa537f3e94aea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 23:33:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6914dae52b354547806ae31d21a77402 : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7bb0a18258eb45db88b2a3288014d507 : SUCCESS in 6m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1df08f77f860445c97b641190a9672de : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ab309456fbb485ab3bc28f38885ff76 : SUCCESS in 5m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/88c5f6e9b647442b9949bd26f02e0afb : SUCCESS in 5m 25s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7fda47e95e0a4927b3af79d95184852d : SUCCESS in 12m 43s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/1d9f08c083f64829873e91c9c7e67f8f : SUCCESS in 12m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a4a4e8bfc70849b398708d86f3144751 : FAILURE in 13m 08s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/ce2966d73e224ee187f77cb7d0ee3a57 : SUCCESS in 11m 13s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/cbfd5ae16f61470a806f8733aee64401 : SUCCESS in 12m 30s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/4b024020e35748219479ab5afc83b139 : SUCCESS in 6m 19s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/3916cdf4267c408fa515c9074b6b0405 : SUCCESS in 6m 42s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/04f3e3c2c2b443b6a59f34f9fbf2f470 : SUCCESS in 11m 10s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/1dfe39264a2f4fcd9148c8b815d5dc85 : SUCCESS in 6m 35s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a9ffaa1a45394652b7b3c710b58de486 : SUCCESS in 18m 27s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/1c4b815138904dcf9b7f0aee19eba7af : SUCCESS in 28m 14s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d588dbf228b24f2386348e2f7d8c6226 : SUCCESS in 14m 52s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/7ec7608ddc7d47d38d320a99a1b04c78 : SUCCESS in 22m 28s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/fc32381f4de24d0285c5f90fb564f9df : SUCCESS in 19m 39s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/86c8b3aec97e4cd3a4ea641a14272aef : SUCCESS in 16m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fbc0991d1a1248adae11faa82ea2a516 : SUCCESS in 12m 28s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/593a5ebd355342a2af98c7d2396992c7 : SUCCESS in 6m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b8f04b37d00d468d9d41635af0cc0612 : SUCCESS in 10m 28s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/72e186405a714493a73de91b2998eb50 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/707d4e7bb3e44cc1b8b190f7151f8fbd : SUCCESS in 4m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/19bc938410084f7da9d7040aa529aa1e : SUCCESS in 1h 13m 52s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/9839f8b5daa94bfabe32fd64d867f525 : SUCCESS in 1h 33m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/cb425a4fd2ba4c289076c82d0dbf2a22 : SUCCESS in 52m 01s","accounts_in_message":[],"_revision_number":2},{"id":"425659fe08cee6c4210428766b51d5cdc96f5e5b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-02 23:58:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"868a6d2544944ad65b37819374a73c03a8a340b1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-03 00:41:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ec054955421e1d7b7337e09060196c8d0f79f315","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 00:42:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"63f9fe001180d7a1ec0352d3493bd254efc9279a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 02:22:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/235cd9a46ca44cce94afb9dd75f5d99d : SUCCESS in 3m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/db59134c652345249fe5bc4f8af1c5ea : SUCCESS in 3m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/523b8db5d1d14d9b9d048ed941789d5c : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e2fdfec494341f6a84ca5eab99192cc : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/da29d87ea3294b21a5d484586af260e1 : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/171b5f00df274dc0aa0acda7109dc9c6 : SUCCESS in 6m 50s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/42c283d6ad9c48a4bf18b4d96f80aec2 : SUCCESS in 8m 37s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8c64300cba1c4b0a9c7d14a776638e78 : SUCCESS in 7m 03s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/8dd032a59e0740ef9e04dbe4aa3f1350 : SUCCESS in 9m 13s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/3a38f41d177848bdb3a7569ad9a41249 : SUCCESS in 4m 46s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/bb470da779d44360956a2de95b3cbd43 : SUCCESS in 4m 52s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/9fffc7976b1b418ba0d3b4e730614e45 : SUCCESS in 7m 00s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/9b0864d7587b454a8f737c288bd755c6 : SUCCESS in 11m 10s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/69d36ccbb61c4c02a2c1af699baedcd0 : SUCCESS in 10m 37s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/368818992cf14c4088732f50ece095bb : SUCCESS in 13m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/005c81f6e00646108b77fdd32bbb1835 : SUCCESS in 14m 15s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/0390294282df42fa8d4d2a1fb1c12962 : SUCCESS in 23m 38s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5f4b33ddf25a40cd8d16423b606ab687 : SUCCESS in 9m 11s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/d46aef90fc4448b38ac5019fabca7fae : SUCCESS in 13m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d1dbde4a58bd42aeb98ee9e95c33ea34 : FAILURE in 12m 18s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/16c1e488fa7743ce8a62ea555adb8659 : SUCCESS in 4m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/7c7dbe1440a144929ac6d5c8875c0001 : SUCCESS in 8m 59s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/53b8138afa464326a399a81d8b6e27bc : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4e1eec024084758b14da31bb6b193e8 : SUCCESS in 4m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e937ac9c057d4f4cb0f8ddfa2232d5b8 : SUCCESS in 59m 03s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/56c60e30d0a543d296648c624c09c177 : SUCCESS in 1h 25m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c95f1616a2094f6abcb38d911c271b9c : SUCCESS in 33m 59s","accounts_in_message":[],"_revision_number":2},{"id":"d2c8a123d3f9fc9479aec2276b82272ccb26a05b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-03 02:31:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"20bb6d48506f8a0e5c583ac5df61cd5042cb3de5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 04:02:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaf1ed68b52641c4aa0dc6bc7cb617dd : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1b96fb49419432ea2200c7eedd73c4e : SUCCESS in 5m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/53954d63bf7344acaaab253793e3b457 : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70b63b9d4f214a43923efd626c0d984b : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/18e5ae6f84a4405a847361d9d722e554 : SUCCESS in 4m 01s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/289139b3b3554ea7a3bcf2a98e43bf0b : SUCCESS in 7m 49s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/b92faa5ddf2d4b89a07a1a70efe554db : SUCCESS in 8m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f7877fafc7104772bb46fa3e2257a39f : FAILURE in 7m 14s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6b1cefced50b476da492ae1f9f371531 : SUCCESS in 8m 29s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/39042ebc97d347be9da747eb9c2fe7da : SUCCESS in 8m 58s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/495acf4d296e4c6b881eeeb062f4ae48 : SUCCESS in 5m 04s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/d7ceaedf6e55478ca894af554f796ade : SUCCESS in 5m 03s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0c387ecc1fcc44588de2eae0183982b1 : SUCCESS in 9m 06s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/cd59e1d2ccad4a84b4b75e35bcc90c3f : SUCCESS in 8m 46s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/f17803703a6940f5a28d20037c4be4de : SUCCESS in 34m 09s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/8ce20a81b1fe46eca82691665d101d74 : SUCCESS in 12m 33s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/f0eaf348acc54f74b4027c27c2b26bdf : SUCCESS in 25m 49s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/b1f1f291ef93497aaee445e9e70346e7 : SUCCESS in 23m 25s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/0e26953b88b7449fafc7e1a50d8ca5b8 : SUCCESS in 12m 02s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/3f9d7647015f49eb9e25ed28cd448442 : SUCCESS in 14m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/58d8daf3f83345f5a8834b8327316636 : SUCCESS in 19m 37s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/e5d76b750fdf445eba552090552fb993 : SUCCESS in 5m 08s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d52a7824b486480884d449e14df38c34 : SUCCESS in 8m 58s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/19154e4ec6d045389fdc87b4270fd6aa : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22f1e59c864c435dafbf21c980637a7b : SUCCESS in 4m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8bd1d4e03b1844c9a8b9c83592fe8308 : FAILURE in 8m 38s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/3474ada3f834459184018f5d59cf191d : SUCCESS in 1h 29m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/85701ab7acf34cf7af7c04a1c0ed5ae6 : SUCCESS in 56m 55s","accounts_in_message":[],"_revision_number":2},{"id":"db0d05087c877a1ef9229136eb5eb8c66651e466","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-03 04:09:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"374540b0883fa6e5c64cf1cf2c12779a731df904","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 05:22:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a05c2f243724d2fb6f8850b3d42659b : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/98944c4970a04acbbe6dbf07717a6850 : SUCCESS in 4m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2cbc8379cb1941a9ac46436745f92689 : SUCCESS in 6m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/679f012ea760431cae7cc7a40e0e8bf5 : SUCCESS in 5m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f0bdd7ae9e38426888e6cab9e9ee3bca : SUCCESS in 3m 24s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b3d729085f484192b7292d59f09f6656 : SUCCESS in 11m 30s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/29008c1f297f4141a61b79432b47986d : SUCCESS in 8m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7ba9b666d15e44c2ac951dd668d7c509 : FAILURE in 8m 04s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/44a9612187b6464eb5f55c5dd9f2b94a : SUCCESS in 13m 15s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/a652819011fd48808064c27c244182e9 : SUCCESS in 9m 03s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cbac5247c85346739815f461d7b82784 : SUCCESS in 4m 53s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/d73fba17ef454fc8978aaab42820e957 : SUCCESS in 5m 42s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/c1076120a43d40ddb19b5046816d95f0 : SUCCESS in 11m 05s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/f536e7e0e9274ab8ae474670f7ee69a6 : SUCCESS in 10m 03s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/10b28b71dad1480d8d18cdc0b40853e1 : SUCCESS in 16m 23s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/cb159aeecda24f879471299c426c5291 : SUCCESS in 20m 45s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1dcc52484f8344b99d4ac7664eb55b28 : SUCCESS in 23m 07s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/3c19c6067a244243937ead903fc51538 : SUCCESS in 26m 52s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/97acda4671f140508bbec86088a050aa : SUCCESS in 18m 21s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/d5ed179e52e846efa08b1fef2a3fe0c4 : SUCCESS in 12m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/18c924b884084e5797f31e87c4732109 : SUCCESS in 20m 59s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/100ab60d90e64b83933d6f648541c1da : SUCCESS in 4m 48s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0889d110427846c0858eaef6fdca5641 : SUCCESS in 8m 51s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/68387709f61d40cbbf5efcea2a4854f6 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb9efe89fce248d58e6ace1bf52211e9 : SUCCESS in 3m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2fc76dfc07a7473496accbdb7f23323b : SUCCESS in 1h 01m 10s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/a17f8b36e37f4862b3545b2a999ae007 : SUCCESS in 1h 11m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c3c8750aa945408ebbf537881f946252 : SUCCESS in 55m 39s","accounts_in_message":[],"_revision_number":2},{"id":"edff645162d608bd0383092915565cddddc22fbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 05:22:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"db8a0c952c0def3e5fa6611bc171a60556d856ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 06:59:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a9f580606965cae07a495fce6f919a6f4b5e936a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 06:59:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d006716c800f44038a797182e3d79d9c : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b0a3c9e52d954e179f76f94260846e56 : SUCCESS in 5m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ad99a2aee1eb4e69a157ff0658747e8f : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b768fe1ab91e4ca796693598c22352b2 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a30ebb75bbdd467585bfe24782d206d9 : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/2d9af840ad9a4874a9f9a7be0a35b6f5 : SUCCESS in 7m 26s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/733f8efb62f94e7dbe5d0d640bf67ced : SUCCESS in 9m 01s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/1cf31ee804234a878a514a4c294c5326 : SUCCESS in 9m 10s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/3db073db2e7c4f44b08b9bd100bf8003 : SUCCESS in 9m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/a8447f23e8314e968cb8bc43fe9c4e55 : SUCCESS in 5m 25s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/9c1e035cdbe2424cb3d755f9d8854111 : SUCCESS in 5m 31s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1e8eb77f152d42a4bef63da193fc32b5 : SUCCESS in 8m 54s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/34acd847378b46c49970d6e83d023a21 : SUCCESS in 9m 11s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2cae4cfaf3f642d8b3c8e42c2d7536de : SUCCESS in 24m 45s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/a0995ccbc9764cf5b3f68b8555eb6d5a : SUCCESS in 21m 12s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/31d521d073f44daaa50afc938424afd5 : SUCCESS in 23m 07s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/b19ec44729cc44a18a9d63b7055e83cc : SUCCESS in 19m 12s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2ff84ca96fae42c1add50e47f72e49a4 : SUCCESS in 14m 27s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/70b281f7a7d0492baff098d3eb24cddf : SUCCESS in 24m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3f94bbeb8f064e148aa34a6b9ef6c096 : SUCCESS in 16m 46s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/1b79d1d3f7d64ea4bf55c4c730dc1d1d : SUCCESS in 4m 36s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c10aeeda1649419b839dace7a41db4d2 : SUCCESS in 7m 54s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/7ba78338409345409abfb61b4f3b69ac : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/507976b4cb3a49079fd6052084c6e457 : SUCCESS in 4m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/82cb1c71d57e439d88636497ede324a0 : SUCCESS in 1h 01m 50s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/bc772093f79f47519b1e057ff8d43a52 : SUCCESS in 1h 30m 44s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fbe74c7076884bba8d915f6e93421ce4 : SUCCESS in 1h 30m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e70b99f0d05adefd12b16f4f8882a9140a21e220","revisions":{"7a979b90f641d129c274f44f6b35767ce93c1b18":{"kind":"REWORK","_number":1,"created":"2020-06-02 21:46:30.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/10/733010/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/733010/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/733010/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/733010/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/733010/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/733010/1"}}},"commit":{"parents":[{"commit":"a2eea81ab8d1f731cda05cf5a547ff90b3686655","subject":"Switch to build-wheel-cache jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2eea81ab8d1f731cda05cf5a547ff90b3686655"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-02 21:46:02.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-02 21:46:29.000000000","tz":-420},"subject":"Pin diskimage-builder \u003c 3.0.0 on python2.7","message":"Pin diskimage-builder \u003c 3.0.0 on python2.7\n\ndiskimage-builder has dropped support for python2.7[1] in version\n3.0.0. We need to pin it to \u003c 3.0.0 for the stable branches that\nstill use python2.7 and diskimage-builder.\n\n[1] https://review.opendev.org/728889\n\nChange-Id: I681172d66891b69b5cd538d0598968262d90e368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a979b90f641d129c274f44f6b35767ce93c1b18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a979b90f641d129c274f44f6b35767ce93c1b18"}]},"branch":"refs/heads/stable/rocky"},"e70b99f0d05adefd12b16f4f8882a9140a21e220":{"kind":"REWORK","_number":2,"created":"2020-06-02 21:54:52.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/10/733010/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/733010/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/733010/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/733010/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/733010/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/733010/2"}}},"commit":{"parents":[{"commit":"a2eea81ab8d1f731cda05cf5a547ff90b3686655","subject":"Switch to build-wheel-cache jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a2eea81ab8d1f731cda05cf5a547ff90b3686655"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-02 21:46:02.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-06-02 21:54:47.000000000","tz":-420},"subject":"Pin diskimage-builder \u003c 3.0.0 on python2.7","message":"Pin diskimage-builder \u003c 3.0.0 on python2.7\n\ndiskimage-builder has dropped support for python2.7[1] in version\n3.0.0. We need to pin it to \u003c 3.0.0 for the stable branches that\nstill use python2.7 and diskimage-builder.\n\n[1] https://review.opendev.org/728889\n\nChange-Id: I681172d66891b69b5cd538d0598968262d90e368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e70b99f0d05adefd12b16f4f8882a9140a21e220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e70b99f0d05adefd12b16f4f8882a9140a21e220"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
