)]}'
{"id":"openstack%2Frequirements~733414","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Ia57901bb77f2e3b8871496b8fe0ea9b0f0828e08","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Ia57901bb77f2e3b8871496b8fe0ea9b0f0828e08","subject":"update constraint for os-brick to new release 2.10.3","status":"MERGED","created":"2020-06-03 22:19:54.000000000","updated":"2020-06-04 11:56:18.000000000","submitted":"2020-06-04 11:56:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733414-1591271778829-ef5efc7f","meta_rev_id":"f752a9323e4933dd2908a0451367a37e70fc624b","_number":733414,"virtual_id_number":733414,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 11:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-03 22:25:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-04 09:58:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-03 22:26:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-04 09:58:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 22:25:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-03 22:26:40.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-06-04 09:58:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-04 11:56:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bdbc424eaf3f1bf09dd007359ea37d42e32eb2e9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-03 22:19:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"931bec2ca4eed5e0ad1e4accaac1f4660738da40","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-03 22:25:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e4424fb448c31480f59310c938bccea1ad446a73","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-03 22:26:40.000000000","message":"Patch Set 1: Code-Review+1\n\nThis addresses OSSN-0086","accounts_in_message":[],"_revision_number":1},{"id":"4822aad39adee8d9e4f0ec4e1432d3688be48fdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 00:09:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09639a2622d0440f8f7ba37d105993b0 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c6ceb7e26de4a8fb1805bf60e9574a2 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44c2aa35bc2d45c2a3cb661962d5070a : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb477751eca84f3a84e6abc37f2afc11 : SUCCESS in 6m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f01a9f486b2145be92225f3ddc74395f : SUCCESS in 6m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/469dc7c7de60458cbf6c3922b72cb945 : SUCCESS in 6m 02s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e9207e4767504a82b49496b050bdd7cf : SUCCESS in 11m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db1332ad5b484ec5955815ee7c842b72 : SUCCESS in 15m 15s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/24e52328354a485c92b7c5dabc60e976 : SUCCESS in 9m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/865ffe5fb26d401ea808394c22b63799 : SUCCESS in 9m 46s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/11aac6e00aad47f288f2fc3d33dc9d12 : SUCCESS in 9m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e672da0e7d6c47d1ac419774774c342f : SUCCESS in 5m 35s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/7ddeb0ef233a469986e39b7a45297c6e : SUCCESS in 11m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ccce6905655a44df9adb769bcaa52481 : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b48525e856c74c0cbb110b79a6e8ff7d : SUCCESS in 12m 45s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/2c0e4f8abe1b47d7991eecd04592ab71 : SUCCESS in 34m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c8e060d7022c46f3a0b4f69d03e1632d : SUCCESS in 22m 40s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/7eff5a01db504dcda0bdb305529135ef : SUCCESS in 22m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2df33f0e728b430b8898ed5baa5eafdb : SUCCESS in 23m 54s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a7b57139047f4f259b3fd3b25befa9a3 : SUCCESS in 15m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f8202240ddda415bb3fa01c74760e1c6 : SUCCESS in 15m 01s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/133240ea7a6e43cdbdc372f440b1c219 : SUCCESS in 18m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bde0f43ce8f044d9aa1e89e9bd6b4749 : SUCCESS in 18m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f11f81cb3097451cb3acd549ce1de960 : SUCCESS in 6m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/287726161cc84ddbbe8da47c64b77c3e : SUCCESS in 9m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/61b2b5957587472f85f211ee523ec8ba : SUCCESS in 9m 52s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b039cc71ad904d47805c0c6608607099 : SUCCESS in 1h 09m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c2095bf02ade4b1c838b48e6d1f04096 : SUCCESS in 1h 44m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3f888929feb541d58a237710688b51b2 : SUCCESS in 41m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a966a75a6ee8085b4c5a6b5f0a9dec1fbf1e3fac","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-04 09:58:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0636d319ec18704a8c93bd56d47273924f0adb55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 09:59:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a55235a9d9cbd85ea23c6bfa7d8964e781cc6a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 11:56:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46a14885cff943aeac5211414de76f5e : SUCCESS in 6m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c3608522155e48b7a014a5d0d26de37b : SUCCESS in 8m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9a02c7812fc49dca707cd9158905289 : SUCCESS in 8m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09f9ff2413174792a1f572a69689d14c : SUCCESS in 8m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80215c41882742cea415e233c62be89b : SUCCESS in 6m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f047c5e50eae466894ceb747e846326b : SUCCESS in 7m 06s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/91df1492d275441da8a4bb815c0b161c : SUCCESS in 10m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b328d7614ac74bd28ee8192dd0c611ff : SUCCESS in 11m 03s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/8ba4df46928046d1b79b422abc47d2df : SUCCESS in 12m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5cbc3cc4085e425e98321837c9cc8414 : SUCCESS in 12m 33s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/b9ede238b12f4efbb9ff0f5a971dca62 : SUCCESS in 7m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/40449a18c0f74f58bc7ab8c017998355 : SUCCESS in 7m 26s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0fee4e9dc2e443b794ed29d657b59dec : SUCCESS in 12m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/543406039e194a459ce1e4fc2420585c : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/95fa9b4e36664cb7b118041fb78d3cdd : SUCCESS in 11m 59s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/5182d825587f4cd7a73dd3815f082c52 : SUCCESS in 50m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/41f3e1681bea4fcba5a72d98d5fa82d2 : SUCCESS in 48m 21s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/1cb7177cd457489c82d14993ae36a91c : SUCCESS in 33m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/79e46a9d6a86455980ce0d7118889a22 : SUCCESS in 32m 33s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f769fa0fa0e24afbb3b071e518e6079a : SUCCESS in 13m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5683bf7798be4b66926b3d372adc63ca : SUCCESS in 15m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/8f0a35e7982d45d3b65a72f83fb47faa : SUCCESS in 13m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/549f6bd5ce254efeaf76d106ae33f78a : SUCCESS in 16m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a38c4968c9284d279961f05a94d18d8c : SUCCESS in 7m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d7d4946b4bc6435ebae8bba385fef6c9 : SUCCESS in 10m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/782dfb77843c46ecafbc79c1d05085f5 : SUCCESS in 10m 46s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dfa9612d1c014b2f9b0ed158665b84fb : SUCCESS in 50m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ceaa22acd0874aa3b12d011b40f720eb : SUCCESS in 1h 29m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2f9aa3ba299b4754bffdbbf317ee0093 : SUCCESS in 57m 13s","accounts_in_message":[],"_revision_number":1},{"id":"539a600920aeac42cbc82edef8d0660f94976f88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 11:56:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ae10f52cec8f6d870c7d862d22c66a64ce74f9f","revisions":{"4ae10f52cec8f6d870c7d862d22c66a64ce74f9f":{"kind":"REWORK","_number":1,"created":"2020-06-03 22:19:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/733414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/733414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/733414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/733414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/733414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/733414/1"}}},"commit":{"parents":[{"commit":"b36186bdd352c5c86d6a5c5151809f840aefdf16","subject":"Pin diskimage-builder \u003c 3.0.0 on python2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b36186bdd352c5c86d6a5c5151809f840aefdf16"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-03 22:19:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-03 22:19:51.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.10.3","message":"update constraint for os-brick to new release 2.10.3\n\nChange-Id: Ia57901bb77f2e3b8871496b8fe0ea9b0f0828e08\nmeta:version: 2.10.3\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: I9742b2411c12b5759e1cdd87a79f8d476aec9a60\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ae10f52cec8f6d870c7d862d22c66a64ce74f9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ae10f52cec8f6d870c7d862d22c66a64ce74f9f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
