)]}'
{"id":"openstack%2Frequirements~733415","triplet_id":"openstack%2Frequirements~stable%2Fstein~Iec64ef88eab0de08ad29a06366bd4c45d4dee404","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"Iec64ef88eab0de08ad29a06366bd4c45d4dee404","subject":"update constraint for os-brick to new release 2.8.5","status":"MERGED","created":"2020-06-03 22:22:15.000000000","updated":"2020-06-04 12:02:21.000000000","submitted":"2020-06-04 12:02:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733415-1591272141806-71d89d23","meta_rev_id":"ff5ccf97aa503d313b8a69975a4424680161cea8","_number":733415,"virtual_id_number":733415,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 12:02:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-03 22:25:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-04 09:58:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-06-03 22:27:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-04 09:58:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 22:25:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-03 22:27:21.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-06-04 09:58:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-04 12:02:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6fea77f6d5bb4795b6dd5b186f8eb4a58b6e8b98","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-03 22:22:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5a57572545e81aea1df43832eca9b52c9603e71","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-03 22:25:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"592546ae1a30ea249af9646aae096761a92b7a3d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2020-06-03 22:27:21.000000000","message":"Patch Set 1: Code-Review+1\n\nThis addresses OSSN-0086","accounts_in_message":[],"_revision_number":1},{"id":"cd8fa15b02d7c3025d827ac717d828ed33186fbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 23:56:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e787325c0f2744adadfaa3e31c761b11 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2eec12642fc0486daae5eb1713803ac4 : SUCCESS in 5m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0f11f934d9294760acd4aead679cc324 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4a5bc3b77644ccdb2aa5e127cca18bc : SUCCESS in 5m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62863cbde8af47ddb46225cfc03f39ba : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d52cd9b5c88a4192a8bf207e9c081361 : SUCCESS in 4m 44s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/e81b9bef300a402f81079538c6a5f44b : SUCCESS in 10m 35s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/f5a5f785e00a42c584cee0492e0a2188 : SUCCESS in 12m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d0836d4ca04542eeaaac032c34bdda56 : SUCCESS in 11m 04s (non-voting)\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/a4fd81e8eb8f4566b7c245a2b437612e : SUCCESS in 10m 24s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/8a9106d44cad436ab36e5da23be48133 : SUCCESS in 15m 26s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/e417718fc7ae471691bc3a9dc79a7dd7 : SUCCESS in 5m 59s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/b813dc59337845cf809d3bbf707bfd8f : SUCCESS in 11m 01s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/91a878d09390446fa8457905e8e90b74 : SUCCESS in 10m 55s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/4dc0f9d2a2b24ce4932a773a29c883f6 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3cfb6c6d63794148bcd1687bd5726c90 : SUCCESS in 9m 53s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/e3525b52e97d4f17826ca8e27f02e122 : SUCCESS in 21m 49s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/68ca699527414c4aa409e009de6d000b : SUCCESS in 33m 20s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/fb17af0a3751498292e5e850a25e1607 : SUCCESS in 29m 09s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/85d83f8e03d14eecbf13ec547f8382be : SUCCESS in 28m 44s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/57da42e33a2e4a6781bba8c009da2a11 : SUCCESS in 12m 34s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/fdc8fa4c9db943438b963b4c4cfae9a7 : SUCCESS in 21m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0291808460e34d43a882f7db25824712 : SUCCESS in 20m 50s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/9a5f43d80ed34bf4b224e6502f94a8c8 : SUCCESS in 5m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/8cabd633f275483c80772e1af04042c0 : SUCCESS in 10m 08s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/291ee1b0f2c348b1adf606bc2cd9be71 : SUCCESS in 8m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/53c7ca4bf2db441ea617e5691c9bc0cb : SUCCESS in 1h 05m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5f941209e99c42ae8942f058c8864c42 : SUCCESS in 1h 26m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2a196322ae294469a24b4b742aa33263 : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":1},{"id":"cd580d0c743719454b5c3f8e7435691a38dd58b2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-04 09:58:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f4310189a8af75f3d2de1c08cc872d2b6b749b36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 09:59:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"45ee9ea42a2b1ae6d6b950f2b86fc2442c2fcb67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:02:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c4f679e8b86436b9235e7b75443fae8 : SUCCESS in 6m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d7721ce707c64abc91e417dc88e1d4ba : SUCCESS in 8m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/13ece6f468dc40978ba57427501f19fb : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c319714db4db45f9822a60e860f9a5b8 : SUCCESS in 7m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf4a8412adc34a779a0c2e6a212831fb : SUCCESS in 7m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/146b8a5f6ebf4322bdd2e0a9fd85438c : SUCCESS in 7m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b4e115fcf9c44f99a3ab34bbf3a8684b : SUCCESS in 10m 36s\n- requirements-tox-py35-check-uc https://zuul.opendev.org/t/openstack/build/79e637c5b1ed44748eb9f0f0f972213c : SUCCESS in 10m 20s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/48857e111b2548158f3976c6be614c94 : SUCCESS in 11m 35s\n- cross-cinder-py35 https://zuul.opendev.org/t/openstack/build/0db59e99ced348feaf7285a9a2645024 : SUCCESS in 11m 37s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/49dd844e956d42a2aa0f40d5030e63ec : SUCCESS in 9m 29s\n- cross-glance-py35 https://zuul.opendev.org/t/openstack/build/952a993f4fa04b638bf3aaa25a1df8b6 : SUCCESS in 8m 02s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/d3e16780d12b4fd39cad9b22cbb67d39 : SUCCESS in 14m 21s\n- cross-horizon-py35 https://zuul.opendev.org/t/openstack/build/b5bbf80f3d054020a59a63ed2ca60338 : SUCCESS in 13m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2ece50d1929a4507a7ac38b16300fd78 : SUCCESS in 11m 24s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c413ee2dc4e44d1d87de80fbe6c2c1b6 : SUCCESS in 30m 53s\n- cross-keystone-py35 https://zuul.opendev.org/t/openstack/build/153ed5db5bf04bddae14be226012058f : SUCCESS in 27m 48s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b534949ce10e45759b11f80c6f7f5bb5 : SUCCESS in 30m 58s\n- cross-neutron-py35 https://zuul.opendev.org/t/openstack/build/79145f648e53467587c73cf8c6dc8d36 : SUCCESS in 26m 17s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f69b8ffdf6ed4ffea58d49e0cc916233 : SUCCESS in 14m 43s\n- cross-nova-py35 https://zuul.opendev.org/t/openstack/build/ea3e4448608a4a8187f02f71b98e1870 : SUCCESS in 15m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/93bf92fce83d405696901775ddcc30c6 : SUCCESS in 24m 19s\n- cross-osc-build-sphinx-docs https://zuul.opendev.org/t/openstack/build/92e229eb1350459a82141d468d2deb35 : SUCCESS in 6m 53s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/0bddde96ca524bed85730ac8c2c5b8c2 : SUCCESS in 11m 14s\n- cross-swift-py35 https://zuul.opendev.org/t/openstack/build/e79eac664f65433a995b516678a1d85c : SUCCESS in 10m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e823dd639e44fd3b1a8e3522de7f6d1 : SUCCESS in 1h 17m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/204b60f10b6e413f8b0b87d6a524010a : SUCCESS in 1h 31m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/41b5702d0c534c21a6255d3ac358429e : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f2bedf520d35a2469b735ca786bc4fdffe6fdff5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:02:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"91f0aa3eca667c12b2c8ac9282486ddd27d7b323","revisions":{"91f0aa3eca667c12b2c8ac9282486ddd27d7b323":{"kind":"REWORK","_number":1,"created":"2020-06-03 22:22:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/733415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/733415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/733415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/733415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/733415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/733415/1"}}},"commit":{"parents":[{"commit":"49449167f14400b5f79cedbb02ea69fe5a0fec29","subject":"Pin diskimage-builder \u003c 3.0.0 on python2.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49449167f14400b5f79cedbb02ea69fe5a0fec29"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-03 22:22:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-03 22:22:13.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.8.5","message":"update constraint for os-brick to new release 2.8.5\n\nChange-Id: Iec64ef88eab0de08ad29a06366bd4c45d4dee404\nmeta:version: 2.8.5\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: I9742b2411c12b5759e1cdd87a79f8d476aec9a60\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91f0aa3eca667c12b2c8ac9282486ddd27d7b323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91f0aa3eca667c12b2c8ac9282486ddd27d7b323"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
