)]}'
{"id":"openstack%2Frequirements~733527","triplet_id":"openstack%2Frequirements~master~I11a27bc2149230850d98f9195de32932d2c56ecf","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I11a27bc2149230850d98f9195de32932d2c56ecf","subject":"update constraint for oslo.i18n to new release 5.0.0","status":"MERGED","created":"2020-06-04 10:45:17.000000000","updated":"2020-06-04 16:37:53.000000000","submitted":"2020-06-04 16:35:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733527-1591288559926-0bd3b327","meta_rev_id":"3011b8055f0d7fcf9ad264988236d648b3474370","_number":733527,"virtual_id_number":733527,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 16:35:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-04 12:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-04 14:31:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-04 14:31:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 12:21:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-04 14:31:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-04 16:35:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e92d8ab5fd218d0fa1d4ad12c791f4f08197d27f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-04 10:45:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6264a7a4af4e70156b87077685d16fe58febf79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:10:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400b5c77fc434f409c677c3e5941e764 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5143a35c9c54df9b27716c9f6ae4055 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad3a5e6ab6ba42458abc57b251d4f9ba : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eeb5acc6ebce4fecb9c40a2377275a3d : SUCCESS in 6m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b711893ea304d0ca0056972b2ca2470 : SUCCESS in 7m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aef4d13f61f04751958faa1b503d3b8b : SUCCESS in 7m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8f92114fbde94b719740115337d4b522 : SUCCESS in 17m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7f0afe366326401cb60b2120702100cf : SUCCESS in 15m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/27333ebc7e054fe690a271d5612e5305 : SUCCESS in 12m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/eb76aed0bea04013b06659d44b950f6f : SUCCESS in 7m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bc742ac1190f4b2a870994cc10f38727 : SUCCESS in 13m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f6dc255a1a6343c7a730948d6f39764b : SUCCESS in 8m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/91e0f4384dc14a4c9b8068f9c4b9d577 : SUCCESS in 23m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b6eac9e5d8854619bdd511eb4755d693 : SUCCESS in 29m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9f98dd6835d9450f8e6ffe0f084e3d9d : SUCCESS in 17m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2e063334cbc640939f831170f73aa998 : SUCCESS in 18m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8c8919d8cdda4df7829a353c02d82a2c : SUCCESS in 6m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b8f70e0d380849db810b79a59c15af99 : SUCCESS in 8m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9a5e5bed7e8b48e79ae9ff834471fbde : SUCCESS in 14m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba3e4a019fac49f9adffe2a55c8571b8 : SUCCESS in 57m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/891d0074be184389bb06da05b330b3e1 : SUCCESS in 1h 15m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ea9ee3d9ed04c9691e2632a479aae70 : SUCCESS in 1h 19m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/6dcee1761f024933b161c9fddac776ac : SUCCESS in 41m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b78c2f28f4d01526dc47242726c0164b5336fd7e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-04 12:21:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"374844ef3d1237c173620979044eca5b8717a525","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-04 14:31:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6bb2af74d71cd9eeb2188f036ac6ca1f580b9b68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 14:31:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8def5a545172230558c54148549afe23d9e952a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 16:35:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b85f2b541d30c8f64904d97d212c7a3c89ad8b55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 16:35:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7fb70c0e8e848669d342a1ddffe7087 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8534ba1860a240f694a9478ea93ef1bc : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76edb1a48a8c43019e1dbffee5c4d748 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4640b95ee2b646b1bc181ce101936202 : SUCCESS in 6m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0daba4686b284fa6956a27c06354e8c5 : SUCCESS in 5m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/911970bb068443faa6a8260a247de817 : SUCCESS in 6m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/32fae3c7c7b24cb3849b0fee565b5120 : SUCCESS in 11m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2f70d53ed9ec420b8ac4ec0d365edb7d : SUCCESS in 16m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c5f9f60f2d464aeaa86235a2e71460d5 : SUCCESS in 12m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/33741028359e44bf9d9d20bda4d3cbb5 : SUCCESS in 11m 49s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7003dd3d752e4a14802e5aee31a4a934 : SUCCESS in 14m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/31c930f8588a4c569f82d2fb6652c5bd : SUCCESS in 12m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/95ab768d5567445993195da727fc4199 : SUCCESS in 21m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3aae07dd189d42ca92af4a6908403860 : SUCCESS in 28m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/98fc45f9b41546ad99d49880547740ec : SUCCESS in 17m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a1ddbd1c76dc47609ac600c633e5fe5d : SUCCESS in 23m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e76acc57672e4e00a8275323a33a909b : SUCCESS in 6m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c1c881c2247549a1b7001d189a702066 : SUCCESS in 9m 28s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ac40b03c9cad4c97aa13e0a64ba8f94a : SUCCESS in 12m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0c440a9942394aafa7e9f876e09a31ae : SUCCESS in 14m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2335100534cf428ba6c681f266b81375 : SUCCESS in 51m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/31914109bf3b4979b060507f8cb18330 : SUCCESS in 1h 14m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e237ab928dc648afaa18aa16a38b9b68 : SUCCESS in 1h 29m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/c0d45d78707c4010b853cdde40a4fc0e : SUCCESS in 1h 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3fbdb5f10ad1daff750af3d0937efd862e30c48b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 16:37:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba79a3fa69054bfca3b53aac6a4aa26f : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1315eca8a165fdfdcc3528cffeb49240bc2a147d","revisions":{"1315eca8a165fdfdcc3528cffeb49240bc2a147d":{"kind":"REWORK","_number":1,"created":"2020-06-04 10:45:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/733527/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/733527/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/733527/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/733527/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/733527/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/733527/1"}}},"commit":{"parents":[{"commit":"aeddfbeac59589aa4fe39c5e74b306a02678536e","subject":"Merge \"update constraint for ironic-lib to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeddfbeac59589aa4fe39c5e74b306a02678536e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:45:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:45:12.000000000","tz":0},"subject":"update constraint for oslo.i18n to new release 5.0.0","message":"update constraint for oslo.i18n to new release 5.0.0\n\nChange-Id: I11a27bc2149230850d98f9195de32932d2c56ecf\nmeta:version: 5.0.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I21a5e8a7ca283f4739bf9ead31a66d54f3d7cf82\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1315eca8a165fdfdcc3528cffeb49240bc2a147d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1315eca8a165fdfdcc3528cffeb49240bc2a147d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
