)]}'
{"id":"openstack%2Frequirements~733530","triplet_id":"openstack%2Frequirements~master~Ief817b484a83910cde8bb2f587e4c75d860a4e05","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ief817b484a83910cde8bb2f587e4c75d860a4e05","subject":"update constraint for oslo.context to new release 3.1.0","status":"MERGED","created":"2020-06-04 10:47:06.000000000","updated":"2020-06-04 17:06:28.000000000","submitted":"2020-06-04 17:06:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733530-1591290388634-d5a688fa","meta_rev_id":"83f8cc2519214b81d20d48587d8f4c75f88c5ce5","_number":733530,"virtual_id_number":733530,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 17:06:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-04 13:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-04 14:31:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-04 14:31:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 13:07:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-04 14:31:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-04 17:06:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"737b7c30a567692acbf3cab56f40bc96fa80eab0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-04 10:47:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2994e0dd46004f100e11b9cceb1c6104e7c69210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:25:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd17b822b2ba4b38a905f6b2bf567afe : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97a67641ccd04e4390be5182d5355bad : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffbbdfbee2f14a8ba6cf36e651f56de6 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15de87c4c49947b78c15134258fdc5f3 : SUCCESS in 6m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f05f4606822f4202baeb4f2f036e7583 : SUCCESS in 5m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5eff6f69599c4643ac9fe846fa2e7aa0 : SUCCESS in 5m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d13e99b2bf9e401c8ed059b0a7cee4d4 : SUCCESS in 10m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0645329d7d494ec78b7b2762aebfa81a : SUCCESS in 12m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/81c174b001744673a4d42069391866a8 : SUCCESS in 14m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0adf8217da2842a08ee19f636babffc7 : SUCCESS in 6m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a1a438e7623f4c128cbac2d456a71e8b : SUCCESS in 13m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/477836d3e1384e688b2611050b45b333 : SUCCESS in 9m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9f10bdab4a2f4855bde69a7424891810 : SUCCESS in 18m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bf86910ee633450e8e4c5864386f3c57 : SUCCESS in 26m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a7976a60d41843e8b9e32af8a3ea3990 : SUCCESS in 16m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e4b5017f78e40e1abcee76c39c587f8 : SUCCESS in 20m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5f028696ce6d43b7a31a37413901026a : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/290939d6957749a4a54763969735f2e7 : SUCCESS in 9m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/65c1466e7a894c4189878336ea87adf2 : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d135d332dba4315827b60ec4dd09be4 : SUCCESS in 52m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b82158b463bc4947a949725dbb227253 : SUCCESS in 1h 18m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb40fb26709a49d18ba86fe3e756ce84 : SUCCESS in 1h 32m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/ed438213c1154eb3b4677491ab040065 : SUCCESS in 1h 10m 34s","accounts_in_message":[],"_revision_number":1},{"id":"57a2e33d07c92a7ff7fd12dafcec770b36a64078","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-04 13:07:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c537461c4e7ced1f70af59a44bad062164c0cf2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-04 14:31:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0890ac99dfde70b5813cd9b516b029ab568b878b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 14:31:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8ae7c11ad3e2dcb9d9ae5a7ec487b07a0464a9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:06:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"49537125683697b117e4a7db87d16cf741ab9e85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:06:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdf5067f98714142843624ceb8e2e718 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21710797594745f4b0da85d6d52e298d : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e17a85d8ceb4476934866be6997c772 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/849502d96e824563bb3371c8e32ca9d2 : SUCCESS in 5m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6be5d74731bb45febc8cc9b56ce9df61 : SUCCESS in 6m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ee65832f0c164f9ab4f81a322656a8c3 : SUCCESS in 6m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/96e3309e72a844eda996e771f1fed64a : SUCCESS in 12m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2da512bd4c16478283bb3fa04638ba59 : SUCCESS in 13m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5d96d6b5dee647269b0838f77068967b : SUCCESS in 12m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9eb14925782446dca65dd54ead8687ab : SUCCESS in 6m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/faa900bc402a45148554a417e57a02ce : SUCCESS in 13m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d43a999c989e4c9d9fcf5c4d3adf57b9 : SUCCESS in 10m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1f68bf2e644941119c1f444c5ed1f016 : SUCCESS in 17m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a36035c186f54b09a47daf9c16843bfe : SUCCESS in 30m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/899dcf5a5fbc470fb0f0419fd7557143 : SUCCESS in 18m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5737961b532548eba714c8cc06485fc1 : SUCCESS in 20m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5b200349475f4748875723636fbe3012 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5e85dc0a9e9447c6a2d5582d53e7a67a : SUCCESS in 12m 00s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d3c2f5a825ea48fab147ded8198c7701 : SUCCESS in 13m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/03ec766583a248019b2494e44d584300 : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d229f015ab884247b875df728288dc1c : SUCCESS in 53m 23s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/35ad7bb6543e4ccba9b34191da2d3b67 : SUCCESS in 1h 06m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b475ddebb1c248d4bbd68033fb955997 : SUCCESS in 1h 30m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/956ca2ed8fd9418c9fa8542fdcf86a20 : SUCCESS in 1h 00m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c7fc5a30a7e053c1db759c04e4d11958ba8889d","revisions":{"9c7fc5a30a7e053c1db759c04e4d11958ba8889d":{"kind":"REWORK","_number":1,"created":"2020-06-04 10:47:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/733530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/733530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/733530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/733530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/733530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/733530/1"}}},"commit":{"parents":[{"commit":"aeddfbeac59589aa4fe39c5e74b306a02678536e","subject":"Merge \"update constraint for ironic-lib to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeddfbeac59589aa4fe39c5e74b306a02678536e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:47:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:47:03.000000000","tz":0},"subject":"update constraint for oslo.context to new release 3.1.0","message":"update constraint for oslo.context to new release 3.1.0\n\nChange-Id: Ief817b484a83910cde8bb2f587e4c75d860a4e05\nmeta:version: 3.1.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I21a5e8a7ca283f4739bf9ead31a66d54f3d7cf82\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c7fc5a30a7e053c1db759c04e4d11958ba8889d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c7fc5a30a7e053c1db759c04e4d11958ba8889d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
