)]}'
{"id":"openstack%2Frequirements~733531","triplet_id":"openstack%2Frequirements~master~I7207c5ad30909d6dcb764216bfbce7df1afeef75","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7207c5ad30909d6dcb764216bfbce7df1afeef75","subject":"update constraint for oslo.versionedobjects to new release 2.1.0","status":"MERGED","created":"2020-06-04 10:47:12.000000000","updated":"2020-06-04 17:06:27.000000000","submitted":"2020-06-04 17:06:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733531-1591290387238-0a49205c","meta_rev_id":"c44ad52ab5df7489d1103ddbe665bc81c4eecd25","_number":733531,"virtual_id_number":733531,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 17:06:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-04 13:07:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-04 14:31:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-04 14:31:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 13:07:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-04 14:31:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-04 17:06:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"175932d1892cf57473c3a5eccef2887ce689148f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-04 10:47:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b84c96296d97eada8d102ab09d5851a1a1e8762f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:25:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba68bea78cf424c819a860a3107d83a : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f95a18ec5435483e9d39c70392969865 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c74cfd8406d45e6a100e67d0983f58d : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/989f00475a9d44b7936526b916669d37 : SUCCESS in 8m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e501f47faa8f4063b1902bdb008c2dee : SUCCESS in 5m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/683ade338a814bdeab08e4c848834f0a : SUCCESS in 9m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ca92a225ad7949b6b6db2c89bb27f20f : SUCCESS in 12m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1edf8e4d62d24995a3619f1853591b66 : SUCCESS in 13m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5592398eaaa84d2b9a40583d479c0234 : SUCCESS in 21m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7d50628202684d1aba8302e0c3f7306e : SUCCESS in 10m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b2dbca4d7bf34976873403354da3afba : SUCCESS in 8m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f65f6672104948d48695bc8ce2ad8cd2 : SUCCESS in 8m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/de1d8258135a49fe8b927709bebf8474 : SUCCESS in 27m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/884828b0c1654efd8716058b9c1182b9 : SUCCESS in 25m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4ba7f49f4e4a4cc4b702bc0522cdc24d : SUCCESS in 17m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78351c682f884727867d53c8dd0de459 : SUCCESS in 20m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1038dc0046ff482a83fdf88457b1f332 : SUCCESS in 7m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/97e13ae24c6745c1ad1b98ed7fbb3a8b : SUCCESS in 7m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/811dec7c7d48462aaf24bf2279aaaa38 : SUCCESS in 17m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be110625d1cc4c82bd8a7a3ba0c7d800 : SUCCESS in 1h 00m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1734aa2deec34caca6d000cb40db7ddc : SUCCESS in 1h 14m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4a317f8f56143d7bb82925effaeeb2e : SUCCESS in 1h 25m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/4d711cacc1f142df85afe47115fb2a62 : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8044423b7edc2eaa69012baa6362535ff40bb53a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-04 13:07:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c646e835a9c8825af7a67c65f7a2be4f1c9a05fd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-04 14:31:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"619b4b0d08a0846dedbda47a5080dda169f8ee21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 14:31:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89c07c39c16cb02c2227f40043e2cd4a272e47a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:06:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4241391dd344af7bb178a8077b1bc2c : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e9d1bd4a0534449b3dff3a94226cae3 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e95534a459684388966f0e6aa0214d64 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/691a9efb21264141ad00938efd4e7deb : SUCCESS in 8m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b2f03bdeb7340da8816a05ef73c2297 : SUCCESS in 5m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/93c68c92891e42658f341307df322084 : SUCCESS in 7m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e72ac986da0a4401871b16d07ec842bd : SUCCESS in 15m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e032ebf016854dd0b2376df95ab29d6d : SUCCESS in 17m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9cb1c0b9df404bd79f9b92875d76d074 : SUCCESS in 12m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6ad5178b70ff4782a91ae3c68b9f1f46 : SUCCESS in 8m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/62db241531484d6080fb4f92b6273930 : SUCCESS in 15m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/27e52e144df54e278144c2c5df410267 : SUCCESS in 13m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a361ec98d66f4ed0831f885a1fe67cbf : SUCCESS in 24m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e61ad6f9cd0e472892aa4d8e78267d2d : SUCCESS in 30m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/40ddff33839b4a9db687af6739287eb5 : SUCCESS in 20m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b3991113cb8e4f20b73b08823f1c3ad4 : SUCCESS in 21m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5bb55f98daa44bd2b2409784874f4a52 : SUCCESS in 7m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38a5842dddbd48f7b8f38b795b3892e5 : SUCCESS in 8m 57s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/95df6e287217491e9bdfa6a0d3affc3e : SUCCESS in 16m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f1a5892321ed44f69680f7afd91841d0 : SUCCESS in 12m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f57e532c3554958a9816201518fe9c9 : SUCCESS in 55m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5ae893dff0e14ab48bf5fdbd3dbfbe8c : SUCCESS in 1h 05m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f099cf4426034566972b0fc1ccdd49e9 : SUCCESS in 1h 27m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/c044784c1a594378bfc23a5ffca417f1 : SUCCESS in 1h 01m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d8b1b0498c00f3877c56281ddb3f774e4a71783a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 17:06:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"575db042ab921e262cd9514e62397ca85981c142","revisions":{"575db042ab921e262cd9514e62397ca85981c142":{"kind":"REWORK","_number":1,"created":"2020-06-04 10:47:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/733531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/733531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/733531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/733531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/733531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/733531/1"}}},"commit":{"parents":[{"commit":"aeddfbeac59589aa4fe39c5e74b306a02678536e","subject":"Merge \"update constraint for ironic-lib to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeddfbeac59589aa4fe39c5e74b306a02678536e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:47:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:47:10.000000000","tz":0},"subject":"update constraint for oslo.versionedobjects to new release 2.1.0","message":"update constraint for oslo.versionedobjects to new release 2.1.0\n\nChange-Id: I7207c5ad30909d6dcb764216bfbce7df1afeef75\nmeta:version: 2.1.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I21a5e8a7ca283f4739bf9ead31a66d54f3d7cf82\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/575db042ab921e262cd9514e62397ca85981c142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/575db042ab921e262cd9514e62397ca85981c142"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
