)]}'
{"id":"openstack%2Frequirements~733534","triplet_id":"openstack%2Frequirements~master~I852aadba47cbc98ebf6e18d7916c4c5a4eef526d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I852aadba47cbc98ebf6e18d7916c4c5a4eef526d","subject":"update constraint for oslotest to new release 4.3.0","status":"MERGED","created":"2020-06-04 10:47:57.000000000","updated":"2020-06-04 16:47:23.000000000","submitted":"2020-06-04 16:45:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733534-1591289141510-1728319f","meta_rev_id":"cee5ac37e84c785a392c9676d165914a661c77cf","_number":733534,"virtual_id_number":733534,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 16:45:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-04 12:21:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-06-04 14:31:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-06-04 14:31:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 12:21:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-04 14:31:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-06-04 16:45:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8eb8895206c43fe6c24439764167ce6092ceb9c1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-06-04 10:47:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a1ada0364f8f0eef6335649ff3496a90e475fe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 12:14:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d53d7b2e212b4f9d84e927d44cc4c57c : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3efe923d9934444dadd72632726a09a3 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcee088e75f14c9985341b8f332a09c1 : SUCCESS in 8m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/061ab7d1e8e746cd9cde4c021ba5f597 : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1812e6a1d30a47e5987ae4d43f60f1c1 : SUCCESS in 7m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/08b54109ce7c4045ab4b64d85146c276 : SUCCESS in 8m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/47fde3e766274be3b1a184c7ea7a6d75 : SUCCESS in 20m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9bd2a87a93504850bf5ba172599fe15e : SUCCESS in 20m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d17aec2f3cc0402bad6401bab204c074 : SUCCESS in 14m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3f253e938e8e44a893f6b727448f0e31 : SUCCESS in 9m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1f3f6906eaa940e2bfe039dc1caa3683 : SUCCESS in 12m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/75670c46447143cc80033259405d98c1 : SUCCESS in 12m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d919d26b2e4f469c8587516c60e4c71e : SUCCESS in 26m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0af51cb747a84fd084b02932be360acb : SUCCESS in 32m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6ef696550bcc44748e003cde5185fce2 : SUCCESS in 15m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eb320cd6b7b945c79c22ab2adc96206f : SUCCESS in 17m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/479916e181cb45218861bd13f84dd11a : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ddf2d7fdf4a1470ebfab511b25f5c48f : SUCCESS in 9m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/836ed4a104fa4b99ba0ffe1698938842 : SUCCESS in 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/556f8b1312f04b32aa61a356a7a2f23c : SUCCESS in 51m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4a7f54ef14049d1b3171fd5d65cc1ae : SUCCESS in 1h 07m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e9e718d33ee4b0bb93c084357592a87 : SUCCESS in 1h 14m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/64ff69c761374b91b877827e93258e9f : SUCCESS in 57m 47s","accounts_in_message":[],"_revision_number":1},{"id":"23e23f56fc3a536bd6769ed3e428be8387827c9c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-04 12:21:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2c981e18c04b2d368497e1c830d4619b55ac0711","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-06-04 14:31:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"44533b0e75877f29529358d0d1cc63671548f3fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 14:31:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09576159476e9a82279e72c0877bc56e7360a565","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 16:45:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aa9afd67c0056edd2a09374e84c683a25ac0b012","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 16:45:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7b0843c722d4d69aeddfa8c32049d2e : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0375e6e069bc489da6f3b089978ba4ae : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd332dfe32ee483a87fc1c3bae3c591f : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3c9f9d0c9644a139b32e7ad52c4d119 : SUCCESS in 7m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5a53215189e144958b867e51b0d507d2 : SUCCESS in 5m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fd9b1d2d62bd4538a0c848aad1b13c77 : SUCCESS in 7m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/45b888aa0673493ebd7010e771957b66 : SUCCESS in 10m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1c70d8efe64c4a038fbf5231ff46089c : SUCCESS in 11m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e630aa6ca0ba43d5ba9e9d3d8d968d4b : SUCCESS in 14m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ef81889632c14757aa129af95555e611 : SUCCESS in 7m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/21be452e90d34f1b8045209119b47212 : SUCCESS in 14m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ecf14fe1725743a693c3e8c282dee624 : SUCCESS in 13m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7839d268b40442139901d9457b5e0028 : SUCCESS in 21m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ccfac88772584720a18ca7ee580de094 : SUCCESS in 29m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/061bc90635aa40899ed533369ddbbdff : SUCCESS in 19m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7a3c2ab76c6546f3876e950aa2885e38 : SUCCESS in 22m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/34f1b404a3224910b5a69d788ba8b93b : SUCCESS in 7m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/54c9da43a5c14a229aa75ae1abc847f4 : SUCCESS in 9m 06s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/deda4d6d1c334dc8bb52ef0de18c2820 : SUCCESS in 14m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9cc78a3a871148b29f73189cf00bc6be : SUCCESS in 14m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99a4956b45a64d1d87cd2b30fb40353f : SUCCESS in 1h 04m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f0f5e5546e064d818962f74b78f059db : SUCCESS in 1h 21m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b41c932727c43af84651d6f08e907b6 : SUCCESS in 1h 38m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/d7bb2535b49b42e08424b5dc594c4078 : SUCCESS in 57m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6887852b1ac87164817bb0be8194eb26c3f12c3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 16:47:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/279d5a840b4d4a7bb2d7a2cc2bfdae6f : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c61d420bbd076c4ea57f7bd052d491d14484f19","revisions":{"9c61d420bbd076c4ea57f7bd052d491d14484f19":{"kind":"REWORK","_number":1,"created":"2020-06-04 10:47:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/34/733534/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/733534/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/733534/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/733534/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/733534/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/733534/1"}}},"commit":{"parents":[{"commit":"aeddfbeac59589aa4fe39c5e74b306a02678536e","subject":"Merge \"update constraint for ironic-lib to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aeddfbeac59589aa4fe39c5e74b306a02678536e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:47:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-06-04 10:47:54.000000000","tz":0},"subject":"update constraint for oslotest to new release 4.3.0","message":"update constraint for oslotest to new release 4.3.0\n\nChange-Id: I852aadba47cbc98ebf6e18d7916c4c5a4eef526d\nmeta:version: 4.3.0\nmeta:diff-start: -\nmeta:series: victoria\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I21a5e8a7ca283f4739bf9ead31a66d54f3d7cf82\nmeta:release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c61d420bbd076c4ea57f7bd052d491d14484f19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c61d420bbd076c4ea57f7bd052d491d14484f19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
